)]}'
{"id":"openstack%2Foslo.log~659289","triplet_id":"openstack%2Foslo.log~master~I7773e41ec48246fb56e7366a359c067788c9f5bb","project":"openstack/oslo.log","branch":"master","topic":"periodical-requirements-check","hashtags":[],"change_id":"I7773e41ec48246fb56e7366a359c067788c9f5bb","subject":"Schedule a periodical check of requirements to catch py2.7 issues quickly","status":"MERGED","created":"2019-05-15 13:05:29.000000000","updated":"2019-06-24 22:24:33.000000000","submitted":"2019-06-24 22:24:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659289-1561415073134-d1910bce","meta_rev_id":"15b38401c778070293ba8b344a1936d0ca063551","_number":659289,"virtual_id_number":659289,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-06-24 22:24:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-06-24 20:28:38.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-05-15 15:01:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":2,"date":"2019-05-23 16:01:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-06-24 20:28:38.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-15 15:01:09.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-05-23 16:01:51.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-06-24 20:28:38.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-06-24 22:24:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1187191e3c38921381fab3e61f5e3273e1cfb01f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-15 13:05:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2580c4b68c71366aee1266a9701feb76b9746d9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 13:05:43.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/oslo.log on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027periodic\u0027][\u0027jobs\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - oslo.log-src-grenade-devstack\n          - oslo.log-jsonformatter\n      gate:\n        jobs:\n          - oslo.log-jsonformatter\n      templates:\n        - check-requirements\n        - lib-forward-testing\n        - lib-forward-testing-python3\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python36-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      periodic:\n          jobs:\n              # NOTE(hberaud) Used to check possible issue on requirements\n              # due to the python 2.7 final countdown and requirements\n              # who remove python 2.7 support. We will check it daily\n              # to see potential errors quickly outside the\n              # standard development lifecycle (feature, fix, etc...)\n              check-requirements\n\n  in \"openstack/oslo.log/.zuul.yaml@master\", line 45, column 3","accounts_in_message":[],"_revision_number":1},{"id":"87bddf418cd309fe234e7e9e27cdad2a96001997","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-15 13:08:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3af8e72754a1afaf6d0a7874e5be11312d50e425","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 13:08:58.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/oslo.log on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027periodic\u0027][\u0027jobs\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - oslo.log-src-grenade-devstack\n          - oslo.log-jsonformatter\n      gate:\n        jobs:\n          - oslo.log-jsonformatter\n      templates:\n        - check-requirements\n        - lib-forward-testing\n        - lib-forward-testing-python3\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python36-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      periodic:\n          jobs:\n              check-requirements\n\n  in \"openstack/oslo.log/.zuul.yaml@master\", line 45, column 3","accounts_in_message":[],"_revision_number":2},{"id":"ff1b3a11d4c2ecca0296b082c65fd3caad0d609b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-15 13:12:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f67a72509ec06c23e1ea9be71f861e0687650434","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 13:13:06.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/oslo.log on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027periodic\u0027][\u0027jobs\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - oslo.log-src-grenade-devstack\n          - oslo.log-jsonformatter\n      gate:\n        jobs:\n          - oslo.log-jsonformatter\n      templates:\n        - check-requirements\n        - lib-forward-testing\n        - lib-forward-testing-python3\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python36-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      periodic:\n        jobs:\n          -check-requirements\n\n  in \"openstack/oslo.log/.zuul.yaml@master\", line 45, column 3","accounts_in_message":[],"_revision_number":3},{"id":"8a7d80da41c873f40b2b0c1a9d506fd8f1b7e51b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-15 13:14:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fa3329a7c8c856bca610fc1d457091d4a4e0e9b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 13:14:46.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/oslo.log on branch master.  The error was:\n\n  Job check-requirements not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - oslo.log-src-grenade-devstack\n          - oslo.log-jsonformatter\n      gate:\n        jobs:\n          - oslo.log-jsonformatter\n      templates:\n        - check-requirements\n        - lib-forward-testing\n        - lib-forward-testing-python3\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python36-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      periodic:\n        jobs:\n          - check-requirements\n\n  in \"openstack/oslo.log/.zuul.yaml@master\", line 45, column 3","accounts_in_message":[],"_revision_number":4},{"id":"9ee263196e779d27769da29f6523a677450aeb53","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-15 13:17:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1bf461758319da9cab7c287a7328ebb3151e740d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 13:17:39.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/oslo.log on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027periodic\u0027][\u0027templates\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - oslo.log-src-grenade-devstack\n          - oslo.log-jsonformatter\n      gate:\n        jobs:\n          - oslo.log-jsonformatter\n      templates:\n        - check-requirements\n        - lib-forward-testing\n        - lib-forward-testing-python3\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python36-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      periodic:\n        templates:\n          - check-requirements\n\n  in \"openstack/oslo.log/.zuul.yaml@master\", line 45, column 3","accounts_in_message":[],"_revision_number":5},{"id":"54316078626020f60efff979736b17984a1fa255","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-15 13:26:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0098819ce2040c24051bed6baf61eafcd5594aac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 13:26:20.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/oslo.log on branch master.  The error was:\n\n  Job check-requirements not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: final.countdown.requirements.check\n      description: |\n          Check issues due to python 2.7 final countdown and requirements\n          updates.\n      parent: check-requirements\n\n  in \"openstack/oslo.log/.zuul.yaml@master\", line 44, column 3","accounts_in_message":[],"_revision_number":6},{"id":"778a1a5a3240f4df8eaef931014995d53e44c745","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-15 13:49:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"745e2354383ace42b934cbd539a19775818b3c86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 13:49:39.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/oslo.log on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027periodic\u0027][\u0027branch\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - oslo.log-src-grenade-devstack\n          - oslo.log-jsonformatter\n      gate:\n        jobs:\n          - oslo.log-jsonformatter\n      templates:\n        - check-requirements\n        - lib-forward-testing\n        - lib-forward-testing-python3\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python36-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      periodic:\n        jobs:\n          - requirements-check\n        branch: master\n\n  in \"openstack/oslo.log/.zuul.yaml@master\", line 45, column 3","accounts_in_message":[],"_revision_number":7},{"id":"3a0b84b7dca88377bdb07bcaf36f73f2e51e3697","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-15 13:50:21.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"39978ab8d89a765a3d77534a20468fdaa24fb136","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-15 15:01:09.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"8412aa8b814717ea4f8b77f5ead04bcb8cf9ccd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 15:25:57.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/89/659289/8/check/tempest-full/31c53d0/ : SUCCESS in 1h 30m 37s\n- tempest-full-py3 http://logs.openstack.org/89/659289/8/check/tempest-full-py3/457e4af/ : SUCCESS in 1h 33m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/659289/8/check/openstack-tox-lower-constraints/ec98cb0/ : SUCCESS in 6m 13s\n- openstack-tox-pep8 http://logs.openstack.org/89/659289/8/check/openstack-tox-pep8/f2af2e1/ : SUCCESS in 6m 10s\n- openstack-tox-py27 http://logs.openstack.org/89/659289/8/check/openstack-tox-py27/c54d99c/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/89/659289/8/check/openstack-tox-py36/fee669f/ : SUCCESS in 4m 26s\n- openstack-tox-docs http://logs.openstack.org/89/659289/8/check/openstack-tox-docs/af962a3/html/ : SUCCESS in 6m 15s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/89/659289/8/check/oslo.log-src-grenade-devstack/452fabf/ : SUCCESS in 56m 58s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/89/659289/8/check/oslo.log-jsonformatter/7fbbcc2/ : SUCCESS in 42m 58s","accounts_in_message":[],"_revision_number":8},{"id":"9f1b179f6e696428b28d1f63a152358a0fb212d9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-23 16:01:51.000000000","message":"Patch Set 8: Code-Review+2\n\nI\u0027m onboard with the idea (hence +2) but where are these reported?","accounts_in_message":[],"_revision_number":8},{"id":"ccfb1402028c526a17a89f613a26b81cd7f54e1f","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-24 20:28:38.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)\n\nLet\u0027s give this a try.","accounts_in_message":[],"_revision_number":8},{"id":"ad8275d7f2d0d8a6b61090be721dd44a995f020d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 20:28:53.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"34b764ea2ba5771650bd61b741805017ed96a071","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 22:24:32.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/89/659289/8/gate/tempest-full/2ae2e57/ : SUCCESS in 1h 54m 32s\n- tempest-full-py3 http://logs.openstack.org/89/659289/8/gate/tempest-full-py3/652e67c/ : SUCCESS in 1h 46m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/659289/8/gate/openstack-tox-lower-constraints/42afa84/ : SUCCESS in 3m 07s\n- openstack-tox-pep8 http://logs.openstack.org/89/659289/8/gate/openstack-tox-pep8/6cdcd33/ : SUCCESS in 2m 46s\n- openstack-tox-py27 http://logs.openstack.org/89/659289/8/gate/openstack-tox-py27/342308c/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/89/659289/8/gate/openstack-tox-py36/b838949/ : SUCCESS in 3m 10s\n- openstack-tox-docs http://logs.openstack.org/89/659289/8/gate/openstack-tox-docs/d058bbe/html/ : SUCCESS in 2m 34s\n- oslo.log-jsonformatter http://logs.openstack.org/89/659289/8/gate/oslo.log-jsonformatter/6e20f4b/ : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":8},{"id":"27687d9cc999b760680dfd9a24eebf5b11e4a3a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 22:24:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"8f9a6a19cf1cdf591c038034350b209c317924b9","revisions":{"59343b63507ecb8a77ad7331ea9115706daf6748":{"kind":"REWORK","_number":1,"created":"2019-05-15 13:05:29.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/659289/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/89/659289/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/1"}}},"commit":{"parents":[{"commit":"1d4c0f42f941294612b9548058dbe39d21047bba","subject":"Cap bandit below 1.6.0 version and update sphinx and limit monotonic.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1d4c0f42f941294612b9548058dbe39d21047bba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:03:27.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:03:27.000000000","tz":120},"subject":"Schedule a periodical check of requirements to catch py2.7 issues quickly","message":"Schedule a periodical check of requirements to catch py2.7 issues quickly\n\nPrepare a periodic job used to check possible issue on requirements\ndue to the python 2.7 final countdown and requirements\nwho remove python 2.7 support. We will check it daily\nto see potential errors quickly outside the\nstandard development lifecycle (feature, fix, etc...)\n\nChange-Id: I7773e41ec48246fb56e7366a359c067788c9f5bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/59343b63507ecb8a77ad7331ea9115706daf6748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/59343b63507ecb8a77ad7331ea9115706daf6748"}]},"branch":"refs/heads/master"},"e24ee63a2767668912839cfaa744de18885c7466":{"kind":"REWORK","_number":2,"created":"2019-05-15 13:08:44.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/659289/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/89/659289/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/2"}}},"commit":{"parents":[{"commit":"1d4c0f42f941294612b9548058dbe39d21047bba","subject":"Cap bandit below 1.6.0 version and update sphinx and limit monotonic.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1d4c0f42f941294612b9548058dbe39d21047bba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:03:27.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:08:33.000000000","tz":120},"subject":"Schedule a periodical check of requirements to catch py2.7 issues quickly","message":"Schedule a periodical check of requirements to catch py2.7 issues quickly\n\nPrepare a periodic job used to check possible issue on requirements\ndue to the python 2.7 final countdown and requirements\nwho remove python 2.7 support. We will check it daily\nto see potential errors quickly outside the\nstandard development lifecycle (feature, fix, etc...)\n\nChange-Id: I7773e41ec48246fb56e7366a359c067788c9f5bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e24ee63a2767668912839cfaa744de18885c7466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e24ee63a2767668912839cfaa744de18885c7466"}]},"branch":"refs/heads/master"},"6751ef6d618dad8b2424f1f5e0f91c6f30fa5d58":{"kind":"REWORK","_number":3,"created":"2019-05-15 13:12:52.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/659289/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/89/659289/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/3"}}},"commit":{"parents":[{"commit":"1d4c0f42f941294612b9548058dbe39d21047bba","subject":"Cap bandit below 1.6.0 version and update sphinx and limit monotonic.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1d4c0f42f941294612b9548058dbe39d21047bba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:03:27.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:12:44.000000000","tz":120},"subject":"Schedule a periodical check of requirements to catch py2.7 issues quickly","message":"Schedule a periodical check of requirements to catch py2.7 issues quickly\n\nPrepare a periodic job used to check possible issue on requirements\ndue to the python 2.7 final countdown and requirements\nwho remove python 2.7 support. We will check it daily\nto see potential errors quickly outside the\nstandard development lifecycle (feature, fix, etc...)\n\nChange-Id: I7773e41ec48246fb56e7366a359c067788c9f5bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6751ef6d618dad8b2424f1f5e0f91c6f30fa5d58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6751ef6d618dad8b2424f1f5e0f91c6f30fa5d58"}]},"branch":"refs/heads/master"},"dcc906d5136584b93e3fd6cb0f79d45f5338fa72":{"kind":"REWORK","_number":4,"created":"2019-05-15 13:14:31.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/659289/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/89/659289/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/4"}}},"commit":{"parents":[{"commit":"1d4c0f42f941294612b9548058dbe39d21047bba","subject":"Cap bandit below 1.6.0 version and update sphinx and limit monotonic.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1d4c0f42f941294612b9548058dbe39d21047bba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:03:27.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:14:23.000000000","tz":120},"subject":"Schedule a periodical check of requirements to catch py2.7 issues quickly","message":"Schedule a periodical check of requirements to catch py2.7 issues quickly\n\nPrepare a periodic job used to check possible issue on requirements\ndue to the python 2.7 final countdown and requirements\nwho remove python 2.7 support. We will check it daily\nto see potential errors quickly outside the\nstandard development lifecycle (feature, fix, etc...)\n\nChange-Id: I7773e41ec48246fb56e7366a359c067788c9f5bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/dcc906d5136584b93e3fd6cb0f79d45f5338fa72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/dcc906d5136584b93e3fd6cb0f79d45f5338fa72"}]},"branch":"refs/heads/master"},"d5fe3b2c8b2aee91b63572ba47843f325ae5a55e":{"kind":"REWORK","_number":5,"created":"2019-05-15 13:17:25.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/659289/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/89/659289/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/5"}}},"commit":{"parents":[{"commit":"1d4c0f42f941294612b9548058dbe39d21047bba","subject":"Cap bandit below 1.6.0 version and update sphinx and limit monotonic.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1d4c0f42f941294612b9548058dbe39d21047bba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:03:27.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:17:15.000000000","tz":120},"subject":"Schedule a periodical check of requirements to catch py2.7 issues quickly","message":"Schedule a periodical check of requirements to catch py2.7 issues quickly\n\nPrepare a periodic job used to check possible issue on requirements\ndue to the python 2.7 final countdown and requirements\nwho remove python 2.7 support. We will check it daily\nto see potential errors quickly outside the\nstandard development lifecycle (feature, fix, etc...)\n\nChange-Id: I7773e41ec48246fb56e7366a359c067788c9f5bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d5fe3b2c8b2aee91b63572ba47843f325ae5a55e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d5fe3b2c8b2aee91b63572ba47843f325ae5a55e"}]},"branch":"refs/heads/master"},"f6901d4f00ee739417eb3a9de2730bafedf828b6":{"kind":"REWORK","_number":6,"created":"2019-05-15 13:26:07.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/659289/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/89/659289/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/6"}}},"commit":{"parents":[{"commit":"1d4c0f42f941294612b9548058dbe39d21047bba","subject":"Cap bandit below 1.6.0 version and update sphinx and limit monotonic.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1d4c0f42f941294612b9548058dbe39d21047bba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:03:27.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:25:58.000000000","tz":120},"subject":"Schedule a periodical check of requirements to catch py2.7 issues quickly","message":"Schedule a periodical check of requirements to catch py2.7 issues quickly\n\nPrepare a periodic job used to check possible issue on requirements\ndue to the python 2.7 final countdown and requirements\nwho remove python 2.7 support. We will check it daily\nto see potential errors quickly outside the\nstandard development lifecycle (feature, fix, etc...)\n\nChange-Id: I7773e41ec48246fb56e7366a359c067788c9f5bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f6901d4f00ee739417eb3a9de2730bafedf828b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f6901d4f00ee739417eb3a9de2730bafedf828b6"}]},"branch":"refs/heads/master"},"05929c4a1c6299a423fe36b287b11aa879b96370":{"kind":"REWORK","_number":7,"created":"2019-05-15 13:49:24.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/659289/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/89/659289/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/7"}}},"commit":{"parents":[{"commit":"1d4c0f42f941294612b9548058dbe39d21047bba","subject":"Cap bandit below 1.6.0 version and update sphinx and limit monotonic.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1d4c0f42f941294612b9548058dbe39d21047bba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:03:27.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:49:16.000000000","tz":120},"subject":"Schedule a periodical check of requirements to catch py2.7 issues quickly","message":"Schedule a periodical check of requirements to catch py2.7 issues quickly\n\nPrepare a periodic job used to check possible issue on requirements\ndue to the python 2.7 final countdown and requirements\nwho remove python 2.7 support. We will check it daily\nto see potential errors quickly outside the\nstandard development lifecycle (feature, fix, etc...)\n\nChange-Id: I7773e41ec48246fb56e7366a359c067788c9f5bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/05929c4a1c6299a423fe36b287b11aa879b96370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/05929c4a1c6299a423fe36b287b11aa879b96370"}]},"branch":"refs/heads/master"},"8f9a6a19cf1cdf591c038034350b209c317924b9":{"kind":"REWORK","_number":8,"created":"2019-05-15 13:50:21.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/659289/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/89/659289/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/89/659289/8"}}},"commit":{"parents":[{"commit":"1d4c0f42f941294612b9548058dbe39d21047bba","subject":"Cap bandit below 1.6.0 version and update sphinx and limit monotonic.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1d4c0f42f941294612b9548058dbe39d21047bba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:03:27.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-15 13:50:12.000000000","tz":120},"subject":"Schedule a periodical check of requirements to catch py2.7 issues quickly","message":"Schedule a periodical check of requirements to catch py2.7 issues quickly\n\nPrepare a periodic job used to check possible issue on requirements\ndue to the python 2.7 final countdown and requirements\nwho remove python 2.7 support. We will check it daily\nto see potential errors quickly outside the\nstandard development lifecycle (feature, fix, etc...)\n\nChange-Id: I7773e41ec48246fb56e7366a359c067788c9f5bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/8f9a6a19cf1cdf591c038034350b209c317924b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/8f9a6a19cf1cdf591c038034350b209c317924b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
