)]}'
{"id":"openstack%2Foslo.log~659755","triplet_id":"openstack%2Foslo.log~stable%2Fstein~I4c2a19fae55e8cfea2d5ae9603a1fb086b5db1bf","project":"openstack/oslo.log","branch":"stable/stein","topic":"stein","hashtags":[],"change_id":"I4c2a19fae55e8cfea2d5ae9603a1fb086b5db1bf","subject":"Cap bandit below 1.6.0 version and update sphinx and limit monotonic.","status":"MERGED","created":"2019-05-17 08:56:29.000000000","updated":"2019-05-23 19:27:52.000000000","submitted":"2019-05-23 19:27:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659755-1558639672843-80094cdc","meta_rev_id":"88191d2422e37192ea1f54238106f3a5c9e22afb","_number":659755,"virtual_id_number":659755,"owner":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-05-23 19:27:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":2,"date":"2019-05-23 16:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-05-23 16:05:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-23 16:05:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":1,"date":"2019-05-23 16:06:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 08:59:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-05-22 22:09:16.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-05-23 16:06:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-23 19:27:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"775a09887620e1667a2ac584cfa481d47dbd53ea","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-17 08:56:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01d1de3b870c435167ca44b6818d5d825d5ae1c0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-17 08:59:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"732a87d18a7cc51a2f43c4fdc9338e197e123657","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 10:45:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/55/659755/1/check/requirements-check/f7ea6a0/ : SUCCESS in 3m 46s\n- tempest-full http://logs.openstack.org/55/659755/1/check/tempest-full/016706a/ : SUCCESS in 1h 47m 12s\n- tempest-full-py3 http://logs.openstack.org/55/659755/1/check/tempest-full-py3/419362b/ : SUCCESS in 1h 24m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/659755/1/check/openstack-tox-lower-constraints/0cf31a7/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/55/659755/1/check/openstack-tox-pep8/3eff7fd/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/55/659755/1/check/openstack-tox-py27/81124c5/ : SUCCESS in 4m 39s\n- openstack-tox-py35 http://logs.openstack.org/55/659755/1/check/openstack-tox-py35/958a58e/ : SUCCESS in 5m 48s\n- openstack-tox-py36 http://logs.openstack.org/55/659755/1/check/openstack-tox-py36/ee21231/ : SUCCESS in 6m 35s\n- openstack-tox-docs http://logs.openstack.org/55/659755/1/check/openstack-tox-docs/9528f3c/html/ : SUCCESS in 4m 09s\n- build-openstack-releasenotes http://logs.openstack.org/55/659755/1/check/build-openstack-releasenotes/f8fa307/html/ : SUCCESS in 5m 55s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/55/659755/1/check/oslo.log-src-grenade-devstack/22f6fda/ : SUCCESS in 55m 44s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/55/659755/1/check/oslo.log-jsonformatter/64e9aef/ : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4bc9fa907a09f2a09f35913521526ff4266dd3d8","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-20 17:17:45.000000000","message":"Patch Set 1:\n\nI think this patch can be merge.","accounts_in_message":[],"_revision_number":1},{"id":"dfc702f11f84d0cebcf73798c8164df9dca37a0f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-21 08:15:59.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b0f258878030b215cd7944c3731b4bf864d93b50","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-22 10:07:52.000000000","message":"Patch Set 1:\n\nApparently it\u0027s blocked.","accounts_in_message":[],"_revision_number":1},{"id":"ce8b702b660e82d05f0bd8057b1dcb56eb77914a","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-05-22 22:09:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f2596449c884aa864f67ede412088fae0e49b6b1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-23 16:05:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e0e61687eb99603b8120342acae2d32929e69c76","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-23 16:06:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"99df585c6f737d34665b0c4c8685796584e3d0e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 17:49:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/55/659755/2/check/requirements-check/6098649/ : SUCCESS in 3m 31s\n- tempest-full http://logs.openstack.org/55/659755/2/check/tempest-full/914be37/ : SUCCESS in 1h 38m 42s\n- tempest-full-py3 http://logs.openstack.org/55/659755/2/check/tempest-full-py3/28e4113/ : SUCCESS in 1h 28m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/659755/2/check/openstack-tox-lower-constraints/9970578/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/55/659755/2/check/openstack-tox-pep8/a3f9f4f/ : SUCCESS in 4m 42s\n- openstack-tox-py27 http://logs.openstack.org/55/659755/2/check/openstack-tox-py27/180f3ed/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/55/659755/2/check/openstack-tox-py35/dd4d0fe/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/55/659755/2/check/openstack-tox-py36/37fcd16/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/55/659755/2/check/openstack-tox-docs/7429210/html/ : SUCCESS in 4m 17s\n- build-openstack-releasenotes http://logs.openstack.org/55/659755/2/check/build-openstack-releasenotes/068b3f5/html/ : SUCCESS in 4m 20s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/55/659755/2/check/oslo.log-src-grenade-devstack/1f79f46/ : SUCCESS in 1h 01m 35s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/55/659755/2/check/oslo.log-jsonformatter/386d9fd/ : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":2},{"id":"07577f9402308082376f009f84a6c0de4910b418","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 17:50:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"952efb6e2d17e42d966d308e3fbeafccb7f69fdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 19:27:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ac4e73d7ace2d9da8f7e916de89f488fc7b7f461","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 19:27:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/55/659755/2/gate/requirements-check/0ffb7da/ : SUCCESS in 3m 16s\n- tempest-full http://logs.openstack.org/55/659755/2/gate/tempest-full/3c344cf/ : SUCCESS in 1h 36m 48s\n- tempest-full-py3 http://logs.openstack.org/55/659755/2/gate/tempest-full-py3/c98154b/ : SUCCESS in 1h 31m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/659755/2/gate/openstack-tox-lower-constraints/61cf5b0/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/55/659755/2/gate/openstack-tox-pep8/69d8a89/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/55/659755/2/gate/openstack-tox-py27/207de82/ : SUCCESS in 4m 47s\n- openstack-tox-py35 http://logs.openstack.org/55/659755/2/gate/openstack-tox-py35/ae0a5c5/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/55/659755/2/gate/openstack-tox-py36/38febad/ : SUCCESS in 4m 18s\n- openstack-tox-docs http://logs.openstack.org/55/659755/2/gate/openstack-tox-docs/8624013/html/ : SUCCESS in 4m 08s\n- build-openstack-releasenotes http://logs.openstack.org/55/659755/2/gate/build-openstack-releasenotes/cd5188f/html/ : SUCCESS in 3m 50s\n- oslo.log-jsonformatter http://logs.openstack.org/55/659755/2/gate/oslo.log-jsonformatter/bf09b82/ : SUCCESS in 39m 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c6158e434540dd84e41da95144ffbece0c323b0f","revisions":{"a70cde77abd86fb4179663310363343d1f6a1f9e":{"kind":"REWORK","_number":1,"created":"2019-05-17 08:56:29.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/55/659755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/55/659755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/55/659755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/55/659755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/55/659755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/55/659755/1"}}},"commit":{"parents":[{"commit":"78dbd58c8c458864ec5b1be4d71dcbb9971eab52","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/78dbd58c8c458864ec5b1be4d71dcbb9971eab52"}]}],"author":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-14 12:32:27.000000000","tz":120},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-17 08:56:06.000000000","tz":120},"subject":"Cap bandit below 1.6.0 version and update sphinx and limit monotonic.","message":"Cap bandit below 1.6.0 version and update sphinx and limit monotonic.\n\nBandit 1.6.0 accidentally changed how to exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nbandit has has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so need to start capping it\nthere as well.\n\nLimit monotonic to python \u003c 3.3 and reflect the changes to\nopenstack/requirements introduced\nin change Ib8c1bf08f5fa7463911602b0df19315907c81e04.\n\nCo-Authored-By: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n\nChange-Id: I4c2a19fae55e8cfea2d5ae9603a1fb086b5db1bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a70cde77abd86fb4179663310363343d1f6a1f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a70cde77abd86fb4179663310363343d1f6a1f9e"}]},"branch":"refs/heads/stable/stein"},"c6158e434540dd84e41da95144ffbece0c323b0f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-23 16:05:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/659755/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/55/659755/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/55/659755/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/55/659755/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/55/659755/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/55/659755/2"}}},"commit":{"parents":[{"commit":"f266cf6ddeb98cf8432fc5fff692fe5fe6c3c6d2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f266cf6ddeb98cf8432fc5fff692fe5fe6c3c6d2"}]}],"author":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-14 12:32:27.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-23 16:04:49.000000000","tz":60},"subject":"Cap bandit below 1.6.0 version and update sphinx and limit monotonic.","message":"Cap bandit below 1.6.0 version and update sphinx and limit monotonic.\n\nBandit 1.6.0 accidentally changed how to exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nbandit has has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so need to start capping it\nthere as well.\n\nLimit monotonic to python \u003c 3.3 and reflect the changes to\nopenstack/requirements introduced\nin change Ib8c1bf08f5fa7463911602b0df19315907c81e04.\n\nCo-Authored-By: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\n\nChange-Id: I4c2a19fae55e8cfea2d5ae9603a1fb086b5db1bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/c6158e434540dd84e41da95144ffbece0c323b0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/c6158e434540dd84e41da95144ffbece0c323b0f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
