)]}'
{"id":"openstack%2Foslo.log~664423","triplet_id":"openstack%2Foslo.log~master~I49047cdc58bb9c7e38d1c56f533b933976f427eb","project":"openstack/oslo.log","branch":"master","topic":"test-zuul-cloner","hashtags":[],"change_id":"I49047cdc58bb9c7e38d1c56f533b933976f427eb","subject":"DNM: Test against zuul-cloner removal patch","status":"ABANDONED","created":"2019-06-10 19:19:18.000000000","updated":"2019-06-10 21:16:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b6518d68d9537f3f47ef90dc74a02206a49bf6cb","_number":664423,"virtual_id_number":664423,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-10 21:08:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 21:08:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8826b7e2363000cde3e64438e14285be372a2f16","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-10 19:19:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6ba1b2820a29668281299f9fca8de075c5259de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 21:08:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/23/664423/1/check/tempest-full/6c3b49d/ : SUCCESS in 1h 43m 17s\n- tempest-full-py3 http://logs.openstack.org/23/664423/1/check/tempest-full-py3/662013c/ : SUCCESS in 1h 34m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/664423/1/check/openstack-tox-lower-constraints/51ab540/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/23/664423/1/check/openstack-tox-pep8/0176be2/ : SUCCESS in 4m 42s\n- openstack-tox-py27 http://logs.openstack.org/23/664423/1/check/openstack-tox-py27/a6ed5d2/ : SUCCESS in 6m 36s\n- openstack-tox-py36 http://logs.openstack.org/23/664423/1/check/openstack-tox-py36/024ff3d/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/23/664423/1/check/openstack-tox-docs/f6cf648/html/ : SUCCESS in 5m 15s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/23/664423/1/check/oslo.log-src-grenade-devstack/00b7aa0/ : SUCCESS in 1h 05m 02s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/23/664423/1/check/oslo.log-jsonformatter/afe4783/ : SUCCESS in 39m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6c9a07b136100d81741fce8360615bb458a21c3d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-10 21:16:19.000000000","message":"Abandoned\n\nLooks good.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa8ba487e06ceabf36a88b29acf538fac76e5593","revisions":{"aa8ba487e06ceabf36a88b29acf538fac76e5593":{"kind":"REWORK","_number":1,"created":"2019-06-10 19:19:18.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/23/664423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/23/664423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/23/664423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/23/664423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/23/664423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/23/664423/1"}}},"commit":{"parents":[{"commit":"8c5317152b0811ef11d3078b67e30b36c83bb6e8","subject":"Avoid tox_install.sh for constraints support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/8c5317152b0811ef11d3078b67e30b36c83bb6e8"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-10 19:19:01.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-10 19:19:01.000000000","tz":0},"subject":"DNM: Test against zuul-cloner removal patch","message":"DNM: Test against zuul-cloner removal patch\n\nDepends-On: https://review.opendev.org/663996\nChange-Id: I49047cdc58bb9c7e38d1c56f533b933976f427eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/aa8ba487e06ceabf36a88b29acf538fac76e5593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/aa8ba487e06ceabf36a88b29acf538fac76e5593"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
