)]}'
{"id":"openstack%2Foslo.log~665204","triplet_id":"openstack%2Foslo.log~master~I4757f93c4194bd821232a2f6cc9abf3be00f15d0","project":"openstack/oslo.log","branch":"master","topic":"rm-incubator","hashtags":[],"change_id":"I4757f93c4194bd821232a2f6cc9abf3be00f15d0","subject":"Remove incubator migration docs","status":"MERGED","created":"2019-06-13 18:20:17.000000000","updated":"2019-07-01 06:00:34.000000000","submitted":"2019-07-01 06:00:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665204-1561960834970-90acf6b2","meta_rev_id":"1bfa6480cf6b88036394f6716a747c9e35ce51c9","_number":665204,"virtual_id_number":665204,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-07-01 06:00:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":2,"date":"2019-07-01 05:52:57.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":2,"date":"2019-06-26 14:52:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":1,"date":"2019-07-01 05:52:57.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 14:52:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-07-01 05:52:57.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-07-01 06:00:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3277fb74505da87a72629132c77146677f377413","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-13 18:20:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9f211cdfb5eba74e16d4e47042850e9ab075597","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 19:29:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/665204/1/check/openstack-tox-pep8/699aaa3/ : SUCCESS in 4m 57s\n- openstack-tox-docs http://logs.openstack.org/04/665204/1/check/openstack-tox-docs/b8a6f51/html/ : SUCCESS in 4m 38s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/04/665204/1/check/oslo.log-src-grenade-devstack/ead4c5b/ : SUCCESS in 49m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b856c27bb847186eed852f31874be571851c4c0b","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-26 14:22:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cdd85fce035ce57912b02d5812f888fee3ec394d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-26 14:52:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"065a152415be08957f24fd4b1dfec79397015147","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 15:48:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/665204/2/check/openstack-tox-pep8/2113adc/ : SUCCESS in 2m 37s\n- openstack-tox-docs http://logs.openstack.org/04/665204/2/check/openstack-tox-docs/777edc0/html/ : SUCCESS in 3m 23s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/04/665204/2/check/oslo.log-src-grenade-devstack/04fcee9/ : SUCCESS in 1h 04m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ae896dcf0791cda4d6638f7ecc597f93c7c2b43","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-07-01 05:52:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"874684ae55326162529e568bf28b3ee3ea06ed0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 05:53:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"09c6e129d0d734adb61bff8e708273e968cbb62f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 06:00:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/665204/2/gate/openstack-tox-pep8/fe1a15a/ : SUCCESS in 2m 34s\n- openstack-tox-docs http://logs.openstack.org/04/665204/2/gate/openstack-tox-docs/57478d6/html/ : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":2},{"id":"698508f791eaae2312811fcb5ac7a54ab2810307","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 06:00:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e7a680e4d0542d21cc4272501f95572f8cd6c01e","revisions":{"31baa1e8dac5d6d8139fdbab1bb76ddd6622d26a":{"kind":"REWORK","_number":1,"created":"2019-06-13 18:20:17.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/04/665204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/04/665204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/04/665204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/04/665204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/04/665204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/04/665204/1"}}},"commit":{"parents":[{"commit":"362fdbbf4b9d14e859500ce4c6079490952065e3","subject":"Fix guidelines w.r.t. translation of log messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/362fdbbf4b9d14e859500ce4c6079490952065e3"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-13 18:18:54.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-13 18:18:54.000000000","tz":0},"subject":"Remove incubator migration docs","message":"Remove incubator migration docs\n\nIncubator logging has been gone for years. Nobody still needs these\ndocs, and on the off chance that I\u0027m wrong they\u0027re still available\nin git.\n\nChange-Id: I4757f93c4194bd821232a2f6cc9abf3be00f15d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/31baa1e8dac5d6d8139fdbab1bb76ddd6622d26a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/31baa1e8dac5d6d8139fdbab1bb76ddd6622d26a"}]},"branch":"refs/heads/master"},"e7a680e4d0542d21cc4272501f95572f8cd6c01e":{"kind":"REWORK","_number":2,"created":"2019-06-26 14:22:13.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/04/665204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/04/665204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/04/665204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/04/665204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/04/665204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/04/665204/2"}}},"commit":{"parents":[{"commit":"791c3612b3e15cf23fb42873989ceeedc99fdd34","subject":"Merge \"Schedule a periodical check of requirements to catch py2.7 issues quickly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/791c3612b3e15cf23fb42873989ceeedc99fdd34"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-13 18:18:54.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-26 14:20:12.000000000","tz":0},"subject":"Remove incubator migration docs","message":"Remove incubator migration docs\n\nIncubator logging has been gone for years. Nobody still needs these\ndocs, and on the off chance that I\u0027m wrong they\u0027re still available\nin git.\n\nChange-Id: I4757f93c4194bd821232a2f6cc9abf3be00f15d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e7a680e4d0542d21cc4272501f95572f8cd6c01e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e7a680e4d0542d21cc4272501f95572f8cd6c01e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
