)]}'
{"id":"openstack%2Foslo.log~666516","triplet_id":"openstack%2Foslo.log~stable%2Fstein~I73e284393ce26c91ea675d5be213ff5e910b4dfe","project":"openstack/oslo.log","branch":"stable/stein","topic":"stein","hashtags":[],"change_id":"I73e284393ce26c91ea675d5be213ff5e910b4dfe","subject":"Modify the constraints url in tox.","status":"MERGED","created":"2019-06-20 08:31:24.000000000","updated":"2019-07-18 16:35:53.000000000","submitted":"2019-07-18 16:35:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666516-1563467753749-4866acf4","meta_rev_id":"446716c51c51e15a9e347ac93bd5761506b34314","_number":666516,"virtual_id_number":666516,"owner":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-07-18 12:46:26.000000000","_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-07-18 16:35:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-07-17 10:45:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":2,"date":"2019-07-17 12:35:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":1,"date":"2019-07-17 12:35:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-17 10:45:23.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-17 12:35:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-07-18 16:35:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1abbab8b0c742d4656836b00230adadb69018261","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-06-20 08:31:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31e9038d93737d3cb7ea6098bd79a9af517281ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 10:17:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/16/666516/1/check/tempest-full/d1c797c/ : SUCCESS in 1h 39m 42s\n- tempest-full-py3 http://logs.openstack.org/16/666516/1/check/tempest-full-py3/736c806/ : SUCCESS in 1h 28m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/666516/1/check/openstack-tox-lower-constraints/5d402fd/ : SUCCESS in 4m 45s\n- openstack-tox-pep8 http://logs.openstack.org/16/666516/1/check/openstack-tox-pep8/cc73c63/ : SUCCESS in 10m 01s\n- openstack-tox-py27 http://logs.openstack.org/16/666516/1/check/openstack-tox-py27/f0e63f0/ : SUCCESS in 3m 56s\n- openstack-tox-py35 http://logs.openstack.org/16/666516/1/check/openstack-tox-py35/ea1e308/ : SUCCESS in 5m 51s\n- openstack-tox-py36 http://logs.openstack.org/16/666516/1/check/openstack-tox-py36/15765df/ : SUCCESS in 4m 52s\n- openstack-tox-docs http://logs.openstack.org/16/666516/1/check/openstack-tox-docs/0192346/ : FAILURE in 7m 59s\n- build-openstack-releasenotes http://logs.openstack.org/16/666516/1/check/build-openstack-releasenotes/9045932/html/ : SUCCESS in 4m 45s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/16/666516/1/check/oslo.log-src-grenade-devstack/d605560/ : SUCCESS in 1h 02m 42s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/16/666516/1/check/oslo.log-jsonformatter/41f836c/ : SUCCESS in 50m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c562bdc602e28bab5f39756c3fdec6dd1f4ff06c","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-06-20 13:57:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"67db5a5a0d0adffabef7814ec63148e851568dd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 15:52:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/16/666516/1/check/tempest-full/740ac99/ : SUCCESS in 1h 53m 19s\n- tempest-full-py3 http://logs.openstack.org/16/666516/1/check/tempest-full-py3/e00354b/ : SUCCESS in 1h 51m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/666516/1/check/openstack-tox-lower-constraints/472d789/ : SUCCESS in 6m 45s\n- openstack-tox-pep8 http://logs.openstack.org/16/666516/1/check/openstack-tox-pep8/f658c63/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/16/666516/1/check/openstack-tox-py27/4798995/ : SUCCESS in 10m 44s\n- openstack-tox-py35 http://logs.openstack.org/16/666516/1/check/openstack-tox-py35/d1d2cf8/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/16/666516/1/check/openstack-tox-py36/7523544/ : SUCCESS in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/16/666516/1/check/openstack-tox-docs/f5ad938/ : FAILURE in 3m 45s\n- build-openstack-releasenotes http://logs.openstack.org/16/666516/1/check/build-openstack-releasenotes/ff55bac/html/ : SUCCESS in 4m 55s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/16/666516/1/check/oslo.log-src-grenade-devstack/15c6361/ : SUCCESS in 1h 05m 07s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/16/666516/1/check/oslo.log-jsonformatter/cfbbdff/ : SUCCESS in 38m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4ac9e9491da9cebc424622b68ba15d4aa62d6cca","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-06-20 22:44:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"109b1fe8a5e5e2a7b7ffaaf971e62d4e1d832d00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 00:29:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/16/666516/2/check/tempest-full/cd219bc/ : SUCCESS in 1h 39m 29s\n- tempest-full-py3 http://logs.openstack.org/16/666516/2/check/tempest-full-py3/5cdbefe/ : SUCCESS in 1h 32m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/666516/2/check/openstack-tox-lower-constraints/5b85a30/ : SUCCESS in 4m 34s\n- openstack-tox-pep8 http://logs.openstack.org/16/666516/2/check/openstack-tox-pep8/6d2fe28/ : FAILURE in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/16/666516/2/check/openstack-tox-py27/97fb35b/ : FAILURE in 5m 37s\n- openstack-tox-py35 http://logs.openstack.org/16/666516/2/check/openstack-tox-py35/3808c70/ : FAILURE in 12m 25s\n- openstack-tox-py36 http://logs.openstack.org/16/666516/2/check/openstack-tox-py36/88d5837/ : FAILURE in 6m 54s\n- openstack-tox-docs http://logs.openstack.org/16/666516/2/check/openstack-tox-docs/0748f36/ : FAILURE in 4m 55s\n- build-openstack-releasenotes http://logs.openstack.org/16/666516/2/check/build-openstack-releasenotes/f056269/html/ : SUCCESS in 4m 54s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/16/666516/2/check/oslo.log-src-grenade-devstack/374927c/ : SUCCESS in 55m 14s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/16/666516/2/check/oslo.log-jsonformatter/9c691c7/ : SUCCESS in 42m 15s","accounts_in_message":[],"_revision_number":2},{"id":"6bb07d6405c5f16b1a5e0cee84fb2e26db9bbec1","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-06-24 08:16:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c4fabcb6c19800b317a61b6ebce7b0b61015dc5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 10:07:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/16/666516/3/check/tempest-full/88286af/ : SUCCESS in 1h 50m 54s\n- tempest-full-py3 http://logs.openstack.org/16/666516/3/check/tempest-full-py3/7af32b1/ : SUCCESS in 1h 38m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/666516/3/check/openstack-tox-lower-constraints/e15251f/ : SUCCESS in 5m 03s\n- openstack-tox-pep8 http://logs.openstack.org/16/666516/3/check/openstack-tox-pep8/a8593cb/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/16/666516/3/check/openstack-tox-py27/0dde420/ : SUCCESS in 5m 11s\n- openstack-tox-py35 http://logs.openstack.org/16/666516/3/check/openstack-tox-py35/d40571b/ : SUCCESS in 5m 25s\n- openstack-tox-py36 http://logs.openstack.org/16/666516/3/check/openstack-tox-py36/cf2e10f/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/16/666516/3/check/openstack-tox-docs/1c40bdf/ : FAILURE in 4m 35s\n- build-openstack-releasenotes http://logs.openstack.org/16/666516/3/check/build-openstack-releasenotes/138e48e/html/ : SUCCESS in 5m 06s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/16/666516/3/check/oslo.log-src-grenade-devstack/62156fb/ : SUCCESS in 1h 00m 12s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/16/666516/3/check/oslo.log-jsonformatter/840da38/ : SUCCESS in 40m 11s","accounts_in_message":[],"_revision_number":3},{"id":"5afe86d503e788e9b7dcf8498e7f5e9fd1950388","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-06-28 08:23:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"79cb50c76c354f66e6a4e229325cc39bbee45681","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 10:12:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/16/666516/3/check/tempest-full/14723bc/ : SUCCESS in 1h 34m 41s\n- tempest-full-py3 http://logs.openstack.org/16/666516/3/check/tempest-full-py3/1580c9c/ : FAILURE in 1h 47m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/666516/3/check/openstack-tox-lower-constraints/21ccaed/ : SUCCESS in 4m 15s\n- openstack-tox-pep8 http://logs.openstack.org/16/666516/3/check/openstack-tox-pep8/522837b/ : SUCCESS in 2m 37s\n- openstack-tox-py27 http://logs.openstack.org/16/666516/3/check/openstack-tox-py27/e0a9b32/ : SUCCESS in 3m 28s\n- openstack-tox-py35 http://logs.openstack.org/16/666516/3/check/openstack-tox-py35/d4e6115/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/16/666516/3/check/openstack-tox-py36/b849699/ : SUCCESS in 3m 06s\n- openstack-tox-docs http://logs.openstack.org/16/666516/3/check/openstack-tox-docs/62f7b85/ : FAILURE in 2m 21s\n- build-openstack-releasenotes http://logs.openstack.org/16/666516/3/check/build-openstack-releasenotes/a1d0241/html/ : SUCCESS in 3m 13s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/16/666516/3/check/oslo.log-src-grenade-devstack/8aa2034/ : SUCCESS in 54m 49s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/16/666516/3/check/oslo.log-jsonformatter/10ad65b/ : SUCCESS in 39m 07s","accounts_in_message":[],"_revision_number":3},{"id":"4ad65b1a50dbe9c6e5c74d6f87cb50955ee6495f","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-07-02 13:41:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8a278102fc5ee1b4a43c53928c8235f55990f87b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 15:21:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/16/666516/3/check/tempest-full/38d3963/ : SUCCESS in 1h 33m 11s\n- tempest-full-py3 http://logs.openstack.org/16/666516/3/check/tempest-full-py3/a49a23e/ : SUCCESS in 1h 30m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/666516/3/check/openstack-tox-lower-constraints/7d2104d/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/16/666516/3/check/openstack-tox-pep8/7b8c05f/ : SUCCESS in 2m 42s\n- openstack-tox-py27 http://logs.openstack.org/16/666516/3/check/openstack-tox-py27/ca18c0c/ : SUCCESS in 3m 06s\n- openstack-tox-py35 http://logs.openstack.org/16/666516/3/check/openstack-tox-py35/256c2d4/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/16/666516/3/check/openstack-tox-py36/0843578/ : SUCCESS in 3m 36s\n- openstack-tox-docs http://logs.openstack.org/16/666516/3/check/openstack-tox-docs/d483db4/ : FAILURE in 2m 41s\n- build-openstack-releasenotes http://logs.openstack.org/16/666516/3/check/build-openstack-releasenotes/744294b/html/ : SUCCESS in 3m 39s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/16/666516/3/check/oslo.log-src-grenade-devstack/067f0cb/ : SUCCESS in 1h 06m 15s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/16/666516/3/check/oslo.log-jsonformatter/7d219b8/ : SUCCESS in 38m 57s","accounts_in_message":[],"_revision_number":3},{"id":"426d2c215e12ab9d0934b7520b1749c407a78c9e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-04 12:38:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c7d46d26427f7898c7e6f5388badc44b513c59f2","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-07-17 08:38:15.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2e70d7f51e8a943d4eb4478d203a736b94041612","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-07-17 08:58:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3089355e91b3c51b952b4ba8727abec88e13f9da","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-17 10:45:23.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"013ea6e15a9240762a4daab5f12c6163642cc8bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 11:05:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/16/666516/4/check/requirements-check/374f03b/ : SUCCESS in 2m 14s\n- tempest-full http://logs.openstack.org/16/666516/4/check/tempest-full/14551c5/ : SUCCESS in 2h 04m 16s\n- tempest-full-py3 http://logs.openstack.org/16/666516/4/check/tempest-full-py3/3baa8c5/ : SUCCESS in 1h 32m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/666516/4/check/openstack-tox-lower-constraints/6c9fd15/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/16/666516/4/check/openstack-tox-pep8/15343bb/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/16/666516/4/check/openstack-tox-py27/2b14928/ : SUCCESS in 2m 45s\n- openstack-tox-py35 http://logs.openstack.org/16/666516/4/check/openstack-tox-py35/2179c89/ : SUCCESS in 3m 06s\n- openstack-tox-py36 http://logs.openstack.org/16/666516/4/check/openstack-tox-py36/b47b196/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/16/666516/4/check/openstack-tox-docs/d8f957e/html/ : SUCCESS in 3m 22s\n- build-openstack-releasenotes http://logs.openstack.org/16/666516/4/check/build-openstack-releasenotes/38e1406/html/ : SUCCESS in 3m 39s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/16/666516/4/check/oslo.log-src-grenade-devstack/6491bb2/ : SUCCESS in 51m 29s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/16/666516/4/check/oslo.log-jsonformatter/bca3265/ : SUCCESS in 33m 55s","accounts_in_message":[],"_revision_number":4},{"id":"ad9ae6a35a1084f37b4a5a1ac46859aa49980882","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-17 12:35:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"251be4b117c770801973f9d27f9683598c57df10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 08:43:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b9dd21d73b231fb0745966e49c8578e9a4c6a577","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 10:55:52.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/16/666516/4/gate/requirements-check/e23ccb5/ : SUCCESS in 3m 01s\n- tempest-full http://logs.openstack.org/16/666516/4/gate/tempest-full/c3bf392/ : TIMED_OUT in 2h 08m 29s\n- tempest-full-py3 http://logs.openstack.org/16/666516/4/gate/tempest-full-py3/b9a19bd/ : SUCCESS in 1h 28m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/666516/4/gate/openstack-tox-lower-constraints/0caf21d/ : SUCCESS in 3m 05s\n- openstack-tox-pep8 http://logs.openstack.org/16/666516/4/gate/openstack-tox-pep8/7d7ef17/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/16/666516/4/gate/openstack-tox-py27/d540bb1/ : SUCCESS in 3m 02s\n- openstack-tox-py35 http://logs.openstack.org/16/666516/4/gate/openstack-tox-py35/cb47c26/ : SUCCESS in 3m 18s\n- openstack-tox-py36 http://logs.openstack.org/16/666516/4/gate/openstack-tox-py36/d95d801/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/16/666516/4/gate/openstack-tox-docs/014d978/html/ : SUCCESS in 2m 48s\n- build-openstack-releasenotes http://logs.openstack.org/16/666516/4/gate/build-openstack-releasenotes/c38c579/html/ : SUCCESS in 3m 37s\n- oslo.log-jsonformatter http://logs.openstack.org/16/666516/4/gate/oslo.log-jsonformatter/4ac288b/ : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":4},{"id":"dfff7af39b31b2a3f6fc7ad8039dbadd46c59203","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-07-18 12:46:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"eac75be3fc6c685f945a98e96bc725c09d0b1c98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 14:35:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/16/666516/4/check/requirements-check/5e6d619/ : SUCCESS in 2m 54s\n- tempest-full http://logs.openstack.org/16/666516/4/check/tempest-full/b04fb06/ : SUCCESS in 1h 47m 36s\n- tempest-full-py3 http://logs.openstack.org/16/666516/4/check/tempest-full-py3/1a84724/ : SUCCESS in 1h 30m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/666516/4/check/openstack-tox-lower-constraints/3b50581/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/16/666516/4/check/openstack-tox-pep8/d7425b3/ : SUCCESS in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/16/666516/4/check/openstack-tox-py27/2f35f6c/ : SUCCESS in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/16/666516/4/check/openstack-tox-py35/8cce3de/ : SUCCESS in 3m 49s\n- openstack-tox-py36 http://logs.openstack.org/16/666516/4/check/openstack-tox-py36/6492b10/ : SUCCESS in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/16/666516/4/check/openstack-tox-docs/8cab6c4/html/ : SUCCESS in 2m 29s\n- build-openstack-releasenotes http://logs.openstack.org/16/666516/4/check/build-openstack-releasenotes/b16c3f4/html/ : SUCCESS in 3m 15s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/16/666516/4/check/oslo.log-src-grenade-devstack/ad1421c/ : SUCCESS in 1h 02m 23s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/16/666516/4/check/oslo.log-jsonformatter/b89c895/ : SUCCESS in 48m 26s","accounts_in_message":[],"_revision_number":4},{"id":"6e5f9f878f32c0790b163cc70b16c507066423b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 14:35:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ee492fd2ea2f2aa6dd93dd70fa8bf977307c8986","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 16:35:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/16/666516/4/gate/requirements-check/03fba5b/ : SUCCESS in 2m 18s\n- tempest-full http://logs.openstack.org/16/666516/4/gate/tempest-full/b795e83/ : SUCCESS in 1h 32m 13s\n- tempest-full-py3 http://logs.openstack.org/16/666516/4/gate/tempest-full-py3/fb4bb6b/ : SUCCESS in 1h 54m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/666516/4/gate/openstack-tox-lower-constraints/af4723b/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/16/666516/4/gate/openstack-tox-pep8/66ecaa6/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/16/666516/4/gate/openstack-tox-py27/7e2554b/ : SUCCESS in 3m 26s\n- openstack-tox-py35 http://logs.openstack.org/16/666516/4/gate/openstack-tox-py35/ac688a9/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/16/666516/4/gate/openstack-tox-py36/bb686f0/ : SUCCESS in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/16/666516/4/gate/openstack-tox-docs/9d3b530/html/ : SUCCESS in 3m 20s\n- build-openstack-releasenotes http://logs.openstack.org/16/666516/4/gate/build-openstack-releasenotes/b4673ed/html/ : SUCCESS in 3m 02s\n- oslo.log-jsonformatter http://logs.openstack.org/16/666516/4/gate/oslo.log-jsonformatter/950bef9/ : SUCCESS in 43m 31s","accounts_in_message":[],"_revision_number":4},{"id":"e05f6bfc5cb82193ff719c06da22fcd29ed0f421","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 16:35:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d5589a7366a60e396880dc2bf3fc0253f5feece6","revisions":{"6a8793433ddb92510b967705201ca2c828128c01":{"kind":"REWORK","_number":1,"created":"2019-06-20 08:31:24.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/16/666516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/16/666516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/1"}}},"commit":{"parents":[{"commit":"44879927ea67a4524c64753abbf0438292465ba7","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/44879927ea67a4524c64753abbf0438292465ba7"}]}],"author":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-23 19:48:41.000000000","tz":120},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-06-20 08:30:52.000000000","tz":120},"subject":"Modify the constraints url in tox.","message":"Modify the constraints url in tox.\n\nUse the hotness for constraints.\n\nChange-Id: I73e284393ce26c91ea675d5be213ff5e910b4dfe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6a8793433ddb92510b967705201ca2c828128c01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6a8793433ddb92510b967705201ca2c828128c01"}]},"branch":"refs/heads/stable/stein"},"1caa9c33df31a2bcae9f32225ef9639892f78705":{"kind":"REWORK","_number":2,"created":"2019-06-20 22:44:45.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/16/666516/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/16/666516/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/2"}}},"commit":{"parents":[{"commit":"44879927ea67a4524c64753abbf0438292465ba7","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/44879927ea67a4524c64753abbf0438292465ba7"}]}],"author":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-23 19:48:41.000000000","tz":120},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-06-20 22:44:13.000000000","tz":120},"subject":"Modify the constraints url in tox.","message":"Modify the constraints url in tox.\n\nUse the hotness for constraints.\n\nChange-Id: I73e284393ce26c91ea675d5be213ff5e910b4dfe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1caa9c33df31a2bcae9f32225ef9639892f78705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1caa9c33df31a2bcae9f32225ef9639892f78705"}]},"branch":"refs/heads/stable/stein"},"ddddf79560011dfbb144c4a653bc99abf4848f30":{"kind":"REWORK","_number":3,"created":"2019-06-24 08:16:18.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/16/666516/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/16/666516/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/3"}}},"commit":{"parents":[{"commit":"44879927ea67a4524c64753abbf0438292465ba7","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/44879927ea67a4524c64753abbf0438292465ba7"}]}],"author":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-23 19:48:41.000000000","tz":120},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-06-24 08:15:57.000000000","tz":120},"subject":"Modify the constraints url in tox.","message":"Modify the constraints url in tox.\n\nUse the hotness for constraints.\n\nChange-Id: I73e284393ce26c91ea675d5be213ff5e910b4dfe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ddddf79560011dfbb144c4a653bc99abf4848f30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ddddf79560011dfbb144c4a653bc99abf4848f30"}]},"branch":"refs/heads/stable/stein"},"d5589a7366a60e396880dc2bf3fc0253f5feece6":{"kind":"REWORK","_number":4,"created":"2019-07-17 08:58:08.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/16/666516/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/16/666516/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/16/666516/4"}}},"commit":{"parents":[{"commit":"44879927ea67a4524c64753abbf0438292465ba7","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/44879927ea67a4524c64753abbf0438292465ba7"}]}],"author":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-23 19:48:41.000000000","tz":120},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-07-17 08:57:48.000000000","tz":120},"subject":"Modify the constraints url in tox.","message":"Modify the constraints url in tox.\n\nUse the hotness for constraints.\n\nChange-Id: I73e284393ce26c91ea675d5be213ff5e910b4dfe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d5589a7366a60e396880dc2bf3fc0253f5feece6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d5589a7366a60e396880dc2bf3fc0253f5feece6"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
