)]}'
{"id":"openstack%2Foslo.log~669230","triplet_id":"openstack%2Foslo.log~master~Ic5fe18b6c0b0b59310b0adbf05c2d6583ff759f4","project":"openstack/oslo.log","branch":"master","topic":"sphinx","hashtags":[],"change_id":"Ic5fe18b6c0b0b59310b0adbf05c2d6583ff759f4","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","status":"MERGED","created":"2019-07-05 01:52:08.000000000","updated":"2019-07-08 03:13:54.000000000","submitted":"2019-07-08 03:13:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669230-1562555634659-eaa21e4c","meta_rev_id":"19575b114eb06fcb4f8019893de99f3e3db9ae51","_number":669230,"virtual_id_number":669230,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-08 03:13:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":2,"date":"2019-07-08 03:07:27.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":2,"date":"2019-07-05 09:24:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":1,"date":"2019-07-08 03:07:27.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 09:24:09.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-08 03:07:27.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-07-08 03:13:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a210bfb0e771b250f791ad1ad8d81219d3686e36","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-05 01:52:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a9a86c0fab667ede932cba3e3c87af3d2d5c5a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 02:46:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/30/669230/1/check/requirements-check/a4c7206/ : SUCCESS in 1m 57s\n- openstack-tox-pep8 http://logs.openstack.org/30/669230/1/check/openstack-tox-pep8/edc1dda/ : SUCCESS in 2m 49s\n- openstack-tox-docs http://logs.openstack.org/30/669230/1/check/openstack-tox-docs/aa68f06/html/ : SUCCESS in 2m 31s\n- build-openstack-releasenotes http://logs.openstack.org/30/669230/1/check/build-openstack-releasenotes/cc47747/html/ : SUCCESS in 2m 50s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/30/669230/1/check/oslo.log-src-grenade-devstack/83bccee/ : SUCCESS in 49m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1ecd5da8b9961cdc5b260cf39fcf410eebd7215","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-05 09:24:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"059e24a42a95b64a7b3fc3e88c3ca04de7727532","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-07-08 03:07:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35ea061d9ac6145984ce4ee43eb1d6083c680ba3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 03:07:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61a70578b86d294e7164f21eb8f2784be01ffaba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 03:13:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7f2cb88921df8d7aeef1807aac33e0d55a7de1e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 03:13:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/30/669230/1/gate/requirements-check/79e0887/ : SUCCESS in 2m 59s\n- openstack-tox-pep8 http://logs.openstack.org/30/669230/1/gate/openstack-tox-pep8/21dcad3/ : SUCCESS in 3m 02s\n- openstack-tox-docs http://logs.openstack.org/30/669230/1/gate/openstack-tox-docs/4cb5d80/html/ : SUCCESS in 3m 31s\n- build-openstack-releasenotes http://logs.openstack.org/30/669230/1/gate/build-openstack-releasenotes/a4a89d9/html/ : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f1227cbd2b17b6587a4cbf3558720ec4736afe4","revisions":{"8f1227cbd2b17b6587a4cbf3558720ec4736afe4":{"kind":"REWORK","_number":1,"created":"2019-07-05 01:52:08.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/30/669230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/30/669230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/30/669230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/30/669230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/30/669230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/30/669230/1"}}},"commit":{"parents":[{"commit":"e7a680e4d0542d21cc4272501f95572f8cd6c01e","subject":"Remove incubator migration docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e7a680e4d0542d21cc4272501f95572f8cd6c01e"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-05 01:51:36.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-05 01:51:40.000000000","tz":480},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\nDepend-On: https://review.opendev.org/#/c/663060/\n\nChange-Id: Ic5fe18b6c0b0b59310b0adbf05c2d6583ff759f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/8f1227cbd2b17b6587a4cbf3558720ec4736afe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/8f1227cbd2b17b6587a4cbf3558720ec4736afe4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
