)]}'
{"id":"openstack%2Foslo.log~679312","triplet_id":"openstack%2Foslo.log~master~Ib214f95abfadd91f85b6cce853057a572ec287f6","project":"openstack/oslo.log","branch":"master","topic":"fluent-complex-obj-fix","hashtags":[],"change_id":"Ib214f95abfadd91f85b6cce853057a572ec287f6","subject":"Serialize complex objects in FluentFormatter","status":"MERGED","created":"2019-08-29 17:56:08.000000000","updated":"2020-12-16 18:53:42.000000000","submitted":"2019-12-10 01:10:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679312-1575940227614-656d7ee0","meta_rev_id":"16e40d118a8a6c1fdf269f9ff33e735fab44dc2a","_number":679312,"virtual_id_number":679312,"owner":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 01:10:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-12-09 15:18:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"value":2,"date":"2019-12-09 15:18:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-12-09 15:18:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"value":1,"date":"2019-12-09 15:18:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-26 18:22:44.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-12-09 15:18:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-12-09 15:18:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-12-10 01:10:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-16 15:27:17.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"}],"messages":[{"id":"da98c0284f8e86833dd4161559111781e7c1355d","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2019-08-29 17:56:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1fe44102a8a8097bac8a634436565fb2e3abb35","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2019-08-29 17:57:11.000000000","message":"Patch Set 1:\n\nI\u0027m not sure if it is a good approach, but it solves the issue with serializing instance object while logging.","accounts_in_message":[],"_revision_number":1},{"id":"64662f594ebe83685dea6c2a2aedde0ae31529c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 19:48:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d4d10f78769e4dfd910f91a5a8807432 : SUCCESS in 1h 46m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ebcb8288e0b47e486933d0f773245e1 : SUCCESS in 1h 37m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1806f0c814c4130ab935436c88ac74f : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c615e0632dea4a13be540dd92e80e731 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/62ebd8fee1b24ac1a2f898ba7fd5347e : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22f26f4a2be64965b21162044ec9f701 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03281020f17942c8933ba008a4c2ac97 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0eb480f53f149b593dc77a03f47c77a : SUCCESS in 5m 20s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/3d2f9ae0778b47bc95853eeee8b16d1b : SUCCESS in 1h 05m 09s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/ce75e07045b849a3a4e68a1e4a12b880 : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":1},{"id":"537f2373ca77280cae943bd8e47dbf1eed688797","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-11-26 18:22:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nSeems fine to me. Would be nice to have a unit test for this situation, but I see we don\u0027t have one for the JSONFormatter either. :-/","accounts_in_message":[],"_revision_number":1},{"id":"e4bf5af1f46d6d1a4f22a71b00d9f2401ada2f96","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-09 15:18:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThere\u0027s some cleanup here, but we can keep it separate in case this needs to be backported","accounts_in_message":[],"_revision_number":1},{"id":"54f14bdb518caff919028a98362ea60ab4bd662a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-09 15:18:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e12f00bbbc10a19999e419654ae3a5b9b3506985","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-09 15:18:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3ba135f574f659a6fe674f10c2aebcc2600aa63c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-09 15:18:50.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"38c1af0a2614583cb539a49956c03cc39591e1bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 17:20:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/fdb8d6c8b01e49a48a4be093926bb26e : SUCCESS in 1h 59m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ce8b40cb8ce4ecf8cfe32cba04df2e2 : SUCCESS in 1h 13m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a217cacd15314e49981424b0e76be141 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43f998720cce4fc3b063041b9980483f : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d244e2301cc6496b86d2b2c67b7eeb36 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c0e80a501a44509adbc4b81576cf40d : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c84ff9d201d4ac5bad6d8c4a59b25c2 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc17f4106973481aa977cac4fc194046 : SUCCESS in 2m 56s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/442b3814c63b4693b330d3782b7f5285 : SUCCESS in 54m 46s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/2317dc5ea5a8497293bec55ba3918d36 : SUCCESS in 41m 33s","accounts_in_message":[],"_revision_number":2},{"id":"1acd2d46fc7a0f424284124bb6a95f47a678cb2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 17:21:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9c122a8ed9a101fcb9f9991dab0432c9aa1bea89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 18:58:06.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3b748b732f7345108eff37c3eb9bea13 : FAILURE in 1h 35m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9102b392e2fd4a6e82543f6f600728c3 : FAILURE in 1h 18m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7289e5e11ba2476ea54f7c252c7c1e5a : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dab60a848d9e432bbe78727e825c5e57 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f101cc0780440dd8cd8a8082e8f4ae1 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a03da5ac212d40a39611a5a677577ae5 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/06e4e4d55d114d79853a8c450a27f080 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abb145794fe549739eb86066973cfea2 : SUCCESS in 4m 24s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/f92d342c92164e7dafc689b0511d5be7 : SUCCESS in 41m 25s","accounts_in_message":[],"_revision_number":2},{"id":"913085d8ba7230e07d2f7e19c571cab27d571245","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-09 22:04:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"093d39ba9b8068d0171749ec66d33510da6e6092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 23:38:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d511af2623e14ecba8dd5b9e46f84638 : SUCCESS in 1h 28m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97195bf4b5c440e5a6d7dbbe4a538a0b : SUCCESS in 1h 31m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edb9dc13fd4f4e37baaaf2563658417c : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc6c89823bfa42e4b4e64ee28be5e424 : SUCCESS in 2m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/960c10cd9dc6483490b481d613f46895 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43464c15c73442f3ab969d4e2c04d663 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b82ebe68ed53451f89003292999229d7 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c759be11f214d209dae2e26e7214548 : SUCCESS in 4m 01s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/7de93c35c6214c499db3191112c10d1a : SUCCESS in 1h 08m 00s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/15b16858640a40cd92ecb35b00dd57b9 : SUCCESS in 45m 49s","accounts_in_message":[],"_revision_number":2},{"id":"4de27556b26da6833cdbd7e7fe2b7af8e1e07f56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 23:38:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f139a60c51e4afb2e47afb0a1c0290c058949fcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 01:10:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9684dd3f8d774e36fee99b4b1a1509207dea16ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 01:10:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/eea5b5134db64641ad1f3c07e13471b8 : SUCCESS in 1h 28m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd4d804090d6412ebb1e602af85e7b5f : SUCCESS in 1h 31m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6add8eb1113947c1a8d782b7aad483b1 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92a331b555f24359a05d1b5ee084789c : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70a237169a3142cf8817ffbdee6ede0e : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80ea10912eff4f978cff294486fc441a : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/718251aa95254b1b9f8584a24321a859 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4801b706eb9041709bcd779c177cc731 : SUCCESS in 4m 25s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/e716279b4523454692ec40cbbc4c61c3 : SUCCESS in 51m 36s","accounts_in_message":[],"_revision_number":2},{"id":"e98f6d9cb8bae1a57b1947655514b6e20f7aa3ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 01:11:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/209f44e1e6dd40fab1670e411b86dccb : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2},{"id":"7c02851fa7ecb2f637156b5d714202a0459dc434","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-12-16 15:27:17.000000000","message":"Patch Set 2:\n\nIf this is a bugfix, shoudn\u0027t it be backported to the older branches?","accounts_in_message":[],"_revision_number":2},{"id":"16e40d118a8a6c1fdf269f9ff33e735fab44dc2a","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-12-16 18:53:42.000000000","message":"Patch Set 2:\n\nYes, this would be valid for backport. Feel free to propose it if you want.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b52c8cd247e9b185d176bfbc4c24c5a2cb35e5be","revisions":{"a0e82f317c1e68ff089ba87b17d32497a4a64d92":{"kind":"REWORK","_number":1,"created":"2019-08-29 17:56:08.000000000","uploader":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"ref":"refs/changes/12/679312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/12/679312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/12/679312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/12/679312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/12/679312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/12/679312/1"}}},"commit":{"parents":[{"commit":"3ff497d1cf0dbb4c78babacc32a02b0b2bacd328","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3ff497d1cf0dbb4c78babacc32a02b0b2bacd328"}]}],"author":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2018-07-20 11:17:43.000000000","tz":180},"committer":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2019-08-29 17:53:58.000000000","tz":120},"subject":"Serialize complex objects in FluentFormatter","message":"Serialize complex objects in FluentFormatter\n\nThis change introduces the serialization with basically json dumps,\nfalling back to repr for non-serializable object. We check the types\nof the top level objects passed to the formatter, as for example\nan object enclosed into dictionary passed to the logger should be a\nmuch more rare case.\n\nCloses-Bug: 1782361\nChange-Id: Ib214f95abfadd91f85b6cce853057a572ec287f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a0e82f317c1e68ff089ba87b17d32497a4a64d92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a0e82f317c1e68ff089ba87b17d32497a4a64d92"}]},"branch":"refs/heads/master"},"b52c8cd247e9b185d176bfbc4c24c5a2cb35e5be":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-09 15:18:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/12/679312/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/12/679312/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/12/679312/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/12/679312/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/12/679312/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/12/679312/2"}}},"commit":{"parents":[{"commit":"1f6336a25a90a32c9fa93d0972a1bdfa13c3e61a","subject":"Merge \"Switch to official Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1f6336a25a90a32c9fa93d0972a1bdfa13c3e61a"}]}],"author":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2018-07-20 11:17:43.000000000","tz":180},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-12-09 15:18:45.000000000","tz":0},"subject":"Serialize complex objects in FluentFormatter","message":"Serialize complex objects in FluentFormatter\n\nThis change introduces the serialization with basically json dumps,\nfalling back to repr for non-serializable object. We check the types\nof the top level objects passed to the formatter, as for example\nan object enclosed into dictionary passed to the logger should be a\nmuch more rare case.\n\nCloses-Bug: 1782361\nChange-Id: Ib214f95abfadd91f85b6cce853057a572ec287f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b52c8cd247e9b185d176bfbc4c24c5a2cb35e5be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b52c8cd247e9b185d176bfbc4c24c5a2cb35e5be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
