)]}'
{"id":"openstack%2Foslo.log~680034","triplet_id":"openstack%2Foslo.log~master~Icf10c2c8b1a5c90602dc0880f838bc9481e9b87e","project":"openstack/oslo.log","branch":"master","topic":"ussuri","hashtags":[],"change_id":"Icf10c2c8b1a5c90602dc0880f838bc9481e9b87e","subject":"Add Ussuri release to versionutils","status":"MERGED","created":"2019-09-04 14:56:36.000000000","updated":"2019-09-18 11:00:30.000000000","submitted":"2019-09-18 10:59:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680034-1568804370654-7b169225","meta_rev_id":"39f079c4d66979dc7f7d2b650dfbc321e232dc1a","_number":680034,"virtual_id_number":680034,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-04 19:54:01.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-09-18 10:59:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-09-18 08:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-09-18 08:51:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 08:51:09.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-09-18 10:59:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f29c4f729fa354f126c4eb80deedb93670364068","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-09-04 14:56:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"508343cc5f22d35a5fccadf75ae54c2378ebd68c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 17:59:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2a91b0750db349369668b92da01c4978 : SUCCESS in 1h 44m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09780ddd8c6e4f3dbe1de8ff347742cb : POST_FAILURE in 1h 22m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/085cbc37bb92483b92987fa451ecc5d2 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdbf5b3321144484b774f3b3773daaa3 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9b5adb4346f43deb470ecd355976230 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/561a8affea0b4e84ab9cd4279ad02535 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/93be4e3c7556483fa8686ee69e7fa171 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a20b9a099954651a5dcbea81f19b592 : SUCCESS in 5m 10s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/50b61551006b4a4c92fed6b2e9347071 : SUCCESS in 58m 05s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/4344d4034bfc4dfdb1d75ff2b1316f0f : SUCCESS in 44m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2645cceade58e4e18422b0ab09ae532003ddbfa3","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-09-04 19:54:01.000000000","message":"Patch Set 1:\n\nrecheck post failure","accounts_in_message":[],"_revision_number":1},{"id":"3f18c0fc70371f79b7a41ce155bce0557d3a658f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 21:56:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/7e58e625afcb4d21b75ad3615d0fc498 : SUCCESS in 1h 58m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/137b99c8b86b45f2beb98b8ef9bc4370 : SUCCESS in 1h 36m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e6d3fa69fa342009da4691945c9d4e2 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d8c5733093d404a94ec647550a35ab3 : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c272213c63154209bd0746a345c5b401 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/204d35abfeda443a9a1c18f024b58770 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c5afccd8c0340d0a7f60f1496eb1c3b : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/941442bf5f2941c3b8165a8e66718ffe : SUCCESS in 5m 29s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/6d2bea491b9b4ef4b377cc97349bbde0 : SUCCESS in 1h 09m 11s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/6a56258fdf864277bd2adb04c77188d9 : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5c4b939a8fbd43449deda5c07ec9b00c8be31ef9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-18 08:51:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTrivial change. Fast approving","accounts_in_message":[],"_revision_number":1},{"id":"8b5f007975829b57a1ecd89589dff52fe1c7545b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:51:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13c071c70aa28a84f37aec0fd44594bbceb13c1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 10:59:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c0e18bf9c1d380e86fe3d091ec31cc066fed19e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 10:59:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/cc027849c63942a5a7b1d25351f6225e : SUCCESS in 1h 42m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25ea6985f4d1464c9f9fbbe45d173d99 : SUCCESS in 1h 33m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2544e3dfd20042b68806001500b74095 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1df0eaad2fdf455caa0c2ade5ba49c3a : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c36d7c21c47e49fdaefed552ba4e9190 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/241ac324987f4ffdad1290985a688dab : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9fc069475c34bb0bc6e4532cf525033 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/185be1e2fff545d48f2addac57c0562f : SUCCESS in 3m 29s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/cd922431ef3541bba96842aaed0d758e : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":1},{"id":"99f543b062c8836965969ec3f24cebcbce49604f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 11:00:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec41457f8d874d29abc76454be645890 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"054df38b8cd8b58f9312b856433a77f5fa03a422","revisions":{"054df38b8cd8b58f9312b856433a77f5fa03a422":{"kind":"REWORK","_number":1,"created":"2019-09-04 14:56:36.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/34/680034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/34/680034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/34/680034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/34/680034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/34/680034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/34/680034/1"}}},"commit":{"parents":[{"commit":"3ff497d1cf0dbb4c78babacc32a02b0b2bacd328","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3ff497d1cf0dbb4c78babacc32a02b0b2bacd328"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-09-04 14:56:13.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-09-04 14:56:13.000000000","tz":0},"subject":"Add Ussuri release to versionutils","message":"Add Ussuri release to versionutils\n\nNow that we have a name we can add it to the list.\n\nChange-Id: Icf10c2c8b1a5c90602dc0880f838bc9481e9b87e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/054df38b8cd8b58f9312b856433a77f5fa03a422"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/054df38b8cd8b58f9312b856433a77f5fa03a422"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
