)]}'
{"id":"openstack%2Foslo.log~698215","triplet_id":"openstack%2Foslo.log~master~Ib61e1fd02a5ec98ddcaed2ce42f049636598870f","project":"openstack/oslo.log","branch":"master","topic":"trivial","hashtags":[],"change_id":"Ib61e1fd02a5ec98ddcaed2ce42f049636598870f","subject":"Always use jsonutils.to_primitive \u0027fallback\u0027 parameter","status":"MERGED","created":"2019-12-10 10:56:24.000000000","updated":"2019-12-20 10:18:14.000000000","submitted":"2019-12-20 10:16:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698215-1576837006068-d3a5649f","meta_rev_id":"3c317d8e3d2621f0c6e2d8d7e4c3a0717cd5d320","_number":698215,"virtual_id_number":698215,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 10:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-12-19 17:48:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-12-20 08:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-12-20 08:42:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-19 17:48:37.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-12-20 08:42:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-12-20 10:16:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a5cac6991237103d1a6e96a3e0c2649de129101c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-10 10:56:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db112bc1f81d2da47056b818ac47ca06e97f0ce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 12:27:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d20789c20bfe4cb9b330adcc6551fb54 : SUCCESS in 1h 29m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb2d11f9f5ea4e08a384e581b7429536 : SUCCESS in 1h 22m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c24b7cf38c664e4c9e1378138287393f : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d50b389a2cb4caba7879bcaabb3db42 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3157d9477bc144b7a6466034d9168fa2 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cfc4b9d703845fb8155ad62c5d0c0c9 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63f37a09e92a4c718a468c5caf9cea97 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417d3a972a3e4aaba2916b24389d00a9 : SUCCESS in 6m 46s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/df01889192a341589f49672d9e3b093f : SUCCESS in 1h 09m 21s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/ae00d1ef915c4e11806b8da7e081051d : SUCCESS in 45m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ed7970607343498abac65555f1f069ce838387ab","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-12-19 17:48:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"031a930a63c7f0c5b87e755c6e76caaa98744ca5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-20 08:42:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f4579423cc35f33c544a5e3e0b8e54d27c787bfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 08:42:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e15ef69d692d46e8130271113a4d9b804181781","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 10:16:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/00212d38e1544be0b2a0aab06407b342 : SUCCESS in 1h 26m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1eb77b4597a744bc877763d6feff5394 : SUCCESS in 1h 32m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17a66e5cad83419aab701c87a225db00 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ea5048a20c149909df49935ccddcc08 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b8c223a891740919669e92082fb7a50 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f376d98d4c84009ad7af41317bdb556 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1eb63a6545b444898c3f44132d8ad1eb : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65fc23cf1b044aeebcbccceaba744f2c : SUCCESS in 3m 47s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/2b4be60777184d9ab0a47a9f300efddf : SUCCESS in 40m 51s","accounts_in_message":[],"_revision_number":1},{"id":"cc77c3c525ee07cb33565c01657a8e7a1670d3d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 10:16:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fd00cbde7a008d1625faef97cdb9154d6a1818e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 10:18:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69428d2e24bf4aeeba36abc536bfe900 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8283aa222dd7f994117f8ff63ade9e0ddd02626","revisions":{"f8283aa222dd7f994117f8ff63ade9e0ddd02626":{"kind":"REWORK","_number":1,"created":"2019-12-10 10:56:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/698215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/15/698215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/15/698215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/15/698215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/15/698215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/15/698215/1"}}},"commit":{"parents":[{"commit":"b52c8cd247e9b185d176bfbc4c24c5a2cb35e5be","subject":"Serialize complex objects in FluentFormatter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b52c8cd247e9b185d176bfbc4c24c5a2cb35e5be"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-10 10:55:14.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-10 10:55:17.000000000","tz":0},"subject":"Always use jsonutils.to_primitive \u0027fallback\u0027 parameter","message":"Always use jsonutils.to_primitive \u0027fallback\u0027 parameter\n\nThis was optionally being enabled to handle the case where we did not\nhave oslo.serialization 2.21.1. Our minimum version is now 2.25.0 so\nthis will never be the case. Remove this optionality logic.\n\nChange-Id: Ib61e1fd02a5ec98ddcaed2ce42f049636598870f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f8283aa222dd7f994117f8ff63ade9e0ddd02626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f8283aa222dd7f994117f8ff63ade9e0ddd02626"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
