)]}'
{"id":"openstack%2Foslo.log~700138","triplet_id":"openstack%2Foslo.log~master~I7269b71668b7d37886bdfd02a188cc9d6fd0c958","project":"openstack/oslo.log","branch":"master","topic":"base_python","hashtags":[],"change_id":"I7269b71668b7d37886bdfd02a188cc9d6fd0c958","subject":"tox: Trivial cleanup","status":"MERGED","created":"2019-12-20 10:03:00.000000000","updated":"2020-01-30 13:24:33.000000000","submitted":"2020-01-30 13:23:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700138-1580390584054-f323dc7b","meta_rev_id":"c873c30e7a62197537db062c157324c573944a1c","_number":700138,"virtual_id_number":700138,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 13:23:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":1,"date":"2020-01-09 05:54:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2020-01-14 11:51:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-13 07:59:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":1,"date":"2020-01-13 09:29:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2020-01-30 11:25:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 05:54:58.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-01-13 07:59:19.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2020-01-13 09:29:58.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2020-01-30 11:25:38.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-30 13:23:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07c6e5102a4e73fc3f4ea858c4500a69424198f9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-20 10:03:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1af3a161bfafbac53c8e4c98c8a0d931acf58c46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 11:53:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4d2c4423173a4c9ab9e9148f54177a47 : SUCCESS in 1h 49m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5c437d11aff4eed8e35422aa7b19d0c : SUCCESS in 1h 44m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f06039898ac64b7c9ce212978bce8251 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/811b3e9d028c42fe8dd524819b28dfab : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4acee0a2583c4f48aceae427c70ec4b2 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74e43cbb27e9402693de061e2ab8e616 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a99ffc94160c4973828399eb5c2c924a : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0af88d0ea1c2486983d3b38ed3c1c689 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9f168398fa34b5bb7817463e1d7ed6c : SUCCESS in 3m 45s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/3720363f7f3544f39ab540cd5ab45b27 : SUCCESS in 55m 37s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/ae677dfc75b64638b1ff5888c3d5afcb : SUCCESS in 42m 11s","accounts_in_message":[],"_revision_number":1},{"id":"6e79f15a6a8ed18963d8c1d0270600d52689e793","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 12:39:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70948bfd4a0294623ac34539ebf3f4a82afe2d59","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-08 18:33:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd54f78b4fc7f1e1af97390d90c5baeadc523bb7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-08 18:34:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e9032bca8765ef8729e0e124119fb6ea33225b23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 20:03:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d99c612fe5b74589b104ac7088d87c92 : SUCCESS in 1h 24m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82e86286b45d483fb4d519c459850fc0 : SUCCESS in 1h 13m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f910a054915042b997eb52c3bbc0bc30 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ba3d2a1de0d43b9be313732d6cb9941 : FAILURE in 6m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/07430f941d4e4a9297ab487d52c82d5d : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69832317b72d43fba72602760db93efa : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a4872013fb94885b774de058547ed29 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5e0ed82f20a406298984c7359447c05 : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5e02a396fbb4419be22b8a9edde698c : SUCCESS in 4m 05s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/0a15903ba14e475e9790b388141eb892 : SUCCESS in 58m 29s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/31bba3dc6def4845a42aab346b22c0c0 : SUCCESS in 36m 07s","accounts_in_message":[],"_revision_number":3},{"id":"9246d9d55d6c4e3375ebacb57d56c8f30801986f","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-09 05:54:58.000000000","message":"Patch Set 3: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"785b750f7ca5c12f237ab281f87102cc012bb463","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 07:30:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/920e95f48e9e4d0fa49d2e9304b921db : SUCCESS in 1h 22m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30ffd00d446a428b9cf4f79da7b07dbe : SUCCESS in 1h 30m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f60e49a95904b4f926385185dc97499 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b420e8f126546a498e64f75872ab4b4 : SUCCESS in 2m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a69f3f866f7a4168823fbe0c2f21e487 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65cfd32a85974ca4a4a2b66bf5bc18e3 : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b37ba4ba9d304397a3be4be7bfaac9ec : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f45fabccbdb748beaefff39d4209f66e : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8829e25e43340a381f144738f2a90f9 : SUCCESS in 4m 08s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/cd77fdcf3d294372a63e1325b5763c27 : SUCCESS in 1h 09m 12s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/cbe908b55eed4c99807a43e534ca4a91 : SUCCESS in 40m 26s","accounts_in_message":[],"_revision_number":3},{"id":"3ec0dbe4007c0a0b0c84c8eb3614616bcc2c49a5","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-13 07:59:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d59c437fa9ccf59c4ca0b9351431947502abf017","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2020-01-13 09:29:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b709c805320ffe3779623090419e9d280ae81677","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-14 11:51:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d8dfdc035122cb2c5e0fe24ac680c0885c7fbc9e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-30 11:25:38.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c48944fa0807162c24f30b38e8d459475ce9a21d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 11:25:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cc2189faf58d859b8a6105386897edaccab52e85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 13:23:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/233992ee6efa43bbb05a4bbe7ab2b463 : SUCCESS in 1h 49m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f33452fe0344bae9f9afc5924af8bde : SUCCESS in 1h 37m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c0268f3faa24250add0748d6d683c82 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0b4709001d5444daac1cf88a2840ece : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2cf6d96e19594981b7bd5a43c44014a1 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/612b8ecca3cb47b58b45af5a68ddfa88 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/754844b9fd9742499d95f8791baf1a98 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76f39a30498e412b848383dc7190e67b : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f2f05de40ae41cb9e545ac0c94b2966 : SUCCESS in 6m 13s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/27cff21c56b34286a34aafdc3491c732 : SUCCESS in 46m 49s","accounts_in_message":[],"_revision_number":3},{"id":"b397a5852f32ce65d0ddaefea1abd8a51fe2880b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 13:23:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2223054626753e5899a3da994491758dd1416a1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 13:24:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d223b4d3895848e68131fc437f2dce73 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba2a2318bcb444c0b2627f2ec38dae47 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d22760ad4924ef56e529bbe6caa173446f2f1bf7","revisions":{"ea2ca479d95f8e53c2c4290c2c313b6cb0d2b73d":{"kind":"REWORK","_number":1,"created":"2019-12-20 10:03:00.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/38/700138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/38/700138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/700138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/700138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/700138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/38/700138/1"}}},"commit":{"parents":[{"commit":"b52c8cd247e9b185d176bfbc4c24c5a2cb35e5be","subject":"Serialize complex objects in FluentFormatter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b52c8cd247e9b185d176bfbc4c24c5a2cb35e5be"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 10:02:29.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 10:02:29.000000000","tz":60},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nMove \u0027basepython\u0027 to the top-level \u0027testenv\u0027.\n\nChange-Id: I7269b71668b7d37886bdfd02a188cc9d6fd0c958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ea2ca479d95f8e53c2c4290c2c313b6cb0d2b73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ea2ca479d95f8e53c2c4290c2c313b6cb0d2b73d"}]},"branch":"refs/heads/master"},"bbd0eb97096cc54659912f38684731a55375cd05":{"kind":"REWORK","_number":2,"created":"2020-01-08 18:33:30.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/38/700138/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/38/700138/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/700138/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/700138/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/700138/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/38/700138/2"}}},"commit":{"parents":[{"commit":"b52c8cd247e9b185d176bfbc4c24c5a2cb35e5be","subject":"Serialize complex objects in FluentFormatter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b52c8cd247e9b185d176bfbc4c24c5a2cb35e5be"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 10:02:29.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-08 18:33:21.000000000","tz":60},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nMove \u0027basepython\u0027 to the top-level \u0027testenv\u0027.\n\nChange-Id: I7269b71668b7d37886bdfd02a188cc9d6fd0c958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/bbd0eb97096cc54659912f38684731a55375cd05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/bbd0eb97096cc54659912f38684731a55375cd05"}]},"branch":"refs/heads/master"},"d22760ad4924ef56e529bbe6caa173446f2f1bf7":{"kind":"REWORK","_number":3,"created":"2020-01-08 18:34:13.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/38/700138/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/38/700138/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/700138/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/700138/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/700138/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/38/700138/3"}}},"commit":{"parents":[{"commit":"b52c8cd247e9b185d176bfbc4c24c5a2cb35e5be","subject":"Serialize complex objects in FluentFormatter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b52c8cd247e9b185d176bfbc4c24c5a2cb35e5be"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 10:02:29.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-08 18:34:02.000000000","tz":60},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nMove \u0027basepython\u0027 to the top-level \u0027testenv\u0027.\n\nChange-Id: I7269b71668b7d37886bdfd02a188cc9d6fd0c958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d22760ad4924ef56e529bbe6caa173446f2f1bf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d22760ad4924ef56e529bbe6caa173446f2f1bf7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
