)]}'
{"id":"openstack%2Foslo.log~702114","triplet_id":"openstack%2Foslo.log~master~Idd5df4c8068810bbbd2bd5058ca38df9f5416a62","project":"openstack/oslo.log","branch":"master","topic":"unittest2","hashtags":[],"change_id":"Idd5df4c8068810bbbd2bd5058ca38df9f5416a62","subject":"Drop use of unittest2","status":"MERGED","created":"2020-01-12 08:37:38.000000000","updated":"2020-01-17 15:15:30.000000000","submitted":"2020-01-17 15:12:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702114-1579273978869-f3637170","meta_rev_id":"ea0a8818097fb595367ae9e10e9060615007b90d","_number":702114,"virtual_id_number":702114,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-13 21:37:37.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-17 15:12:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-14 11:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-01-17 13:35:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-01-17 13:35:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 11:58:00.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-17 13:35:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-17 15:12:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57bfabced002ccd45f69ec1e383a4afbe52ccba6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-12 08:37:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03f26e01db4d831877ebb8e9855207f2e7c800a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 08:58:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/875840cf73014a5e85f43026f58932bd : SUCCESS in 2m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/09338dee6f4a48648bc11334c3ce77db : FAILURE in 13m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b0d98ad73b64681b544c8946665394a : FAILURE in 8m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f735a4680041460f91b6a817efb44bdd : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d57c15d067ea43de9b66aa15da1fca93 : SUCCESS in 2m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74fb5eec648b49c9be09a633134eb34e : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67204617324e4e9dbe15071ae4279ed9 : SUCCESS in 2m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1033ddaa581b4a39aa1b587d24c39873 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56ddad0ace7f444fbe874907129837bc : SUCCESS in 2m 55s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/0a30c885e5114d5e9f50837f961bc9fc : FAILURE in 9m 56s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/1d43ea9080ef47bf9b6afd780bb79ef4 : FAILURE in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"97083a66ae2fee99d9274b9d908197b66a874ede","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-12 17:37:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a655cc0d7fb48c572ee3d0275a77e02c9bca2fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 17:55:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/89de27a5c1c542c0a6732a2e723646b9 : SUCCESS in 2m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f82ba5574aa649149bc1279f55c16a31 : FAILURE in 11m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85e2ea852e5246b388220b9de747bb8e : FAILURE in 12m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbb31ee827634c999c11765177384e2a : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9efeaf3602d4ff59c96dc8691988625 : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b8edc8450974df9a9ea21e6fde1f393 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4b68fe53d4449708ac475a2936bb240 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc5f5f928835476990287924c9db4661 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4f2b86c05d44dc86dd70d8cb35def6 : SUCCESS in 3m 18s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/cb8ee1a8e4e34cd8ab47adf0d2a71580 : FAILURE in 4m 11s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/b2f97cc2cd3c4efe9e4880f34e19053f : FAILURE in 11m 01s","accounts_in_message":[],"_revision_number":1},{"id":"954c65a7447fafb110eb73cef2eaf79abc5f3c2e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-13 21:37:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7c0ca093eb444813bf0c9877608bc42d75ea7455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 23:15:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/43fd8868d8b64017a3a55ce96ea06dd7 : SUCCESS in 2m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/01dade7f593747db990676816c71011b : SUCCESS in 1h 37m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f6aa339600044ada6185035b486abd1 : SUCCESS in 1h 24m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e8474082aa6443b8dde014df6bc9ea0 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7341e9f31c841e9b56820edadd37032 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69baa4b11f824f1dae0e6a577467709c : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb51eb35d8214de4a0b0b79ed594fea6 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0f8e09d5f304bcc8de6601194b2fe24 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1058478eb524d5cbc17ecb7d3a649c5 : SUCCESS in 6m 29s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/9e934ce7f78d4b0eac0d3dd532f16a52 : SUCCESS in 1h 01m 54s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/88687338f983453c8e272c6c6b995ef4 : SUCCESS in 38m 14s","accounts_in_message":[],"_revision_number":1},{"id":"849fcbbc6f099d9b43c5434b24aa356659285dab","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-14 11:58:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d32b6369c6e38d9d2927831cf4d3bf44dab38413","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-17 13:35:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2732b4a54224b9621e28cb97fed488066e257637","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 13:35:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89fae7ccbc0ba4d3b2c1e08d944839cebdf9a33f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 15:12:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8e8e8777cca4949f8236acd626c179462bde8b51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 15:12:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca281ea4d7de4e8b8e9237b3db395659 : SUCCESS in 2m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d2608fdb602849c58f3cac2df4e0ac39 : SUCCESS in 1h 32m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d072007b96334e36a3a4264b26a640c7 : SUCCESS in 1h 24m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77b1a85164ec43a3ac7da34e70bfb589 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506f83bb1d7d47ff81f3040b3f8e97f2 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe59361b577c4dbdb32d07156caa074d : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db11a30067234ddda99401364ccb1bed : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d100f75812e34dc1be19090a0872f65e : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d190cafb99244cc89a9b8520bbf006dd : SUCCESS in 4m 09s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/de956f88ed27446bb8aa340884132303 : SUCCESS in 52m 33s","accounts_in_message":[],"_revision_number":1},{"id":"8c23b734e8b0c16654e43d0e4e09184d6952aafe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 15:15:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af51d9e4724494bbdd2badd16bb6e40 : SUCCESS in 2m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"80a6e26650302329fa9aff26efa389bd99e397c5","revisions":{"80a6e26650302329fa9aff26efa389bd99e397c5":{"kind":"REWORK","_number":1,"created":"2020-01-12 08:37:38.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/14/702114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/14/702114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/14/702114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/14/702114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/14/702114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/14/702114/1"}}},"commit":{"parents":[{"commit":"f8283aa222dd7f994117f8ff63ade9e0ddd02626","subject":"Always use jsonutils.to_primitive \u0027fallback\u0027 parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f8283aa222dd7f994117f8ff63ade9e0ddd02626"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-01-12 08:36:28.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-01-12 08:36:28.000000000","tz":-360},"subject":"Drop use of unittest2","message":"Drop use of unittest2\n\nunittest2 was needed for Python version \u003c\u003d 2.6, so it has not been\nneeded for quite some time now. See unittest2 note on:\n\nhttps://docs.python.org/2.7/library/unittest.html\n\nThis replaces unittest2 in favor of the standard unittest package.\n\nChange-Id: Idd5df4c8068810bbbd2bd5058ca38df9f5416a62\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/80a6e26650302329fa9aff26efa389bd99e397c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/80a6e26650302329fa9aff26efa389bd99e397c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
