)]}'
{"id":"openstack%2Foslo.log~705181","triplet_id":"openstack%2Foslo.log~master~I2a5a8b83d965ab6837ae01640ea4525284a0e7be","project":"openstack/oslo.log","branch":"master","topic":"ignore_rules","hashtags":[],"change_id":"I2a5a8b83d965ab6837ae01640ea4525284a0e7be","subject":"Ignore releasenote artifacts files.","status":"MERGED","created":"2020-01-31 09:55:52.000000000","updated":"2020-01-31 13:10:41.000000000","submitted":"2020-01-31 13:08:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705181-1580476138131-04bc3352","meta_rev_id":"dfe46257f3accedcf3536bd235840b6dfaa84e21","_number":705181,"virtual_id_number":705181,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-31 13:08:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-01-31 10:27:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-01-31 10:27:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-31 10:27:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-31 13:08:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"235c07705a7c0fdc86526cc8ae5561077529788f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-31 09:55:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22f60b1586393710f9d5ccf9a2cc8319be3a442c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-31 10:27:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"010769a4902380d6dc3dc85b895feb59af78c5c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 11:31:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1224af42f3d41e69801ad667039fc27 : SUCCESS in 1h 31m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/640b684fe9e54938b37959645e846234 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af0e7d63c2e84d60a4b8c7477b3fc088 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aea9f4bcc1f4474c83c9db706daa2ead : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/418961fc297d4ddc80437a633b554a6d : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82002acacc044d05b9c4b7a16269cf70 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/016f9ccd58864acc80a4c9bc2ab10450 : SUCCESS in 7m 02s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/684d2e3a12b045ac97df25d26802d31f : FAILURE in 1h 05m 54s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/34d64fb164c24b3e9a49ce1ae74c3081 : SUCCESS in 37m 15s","accounts_in_message":[],"_revision_number":1},{"id":"880246a5c8c6f45edaa40582ad52411ec5ae3af0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 11:31:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b353cc36a2eafb03115c1e0b1097ca1f01be54e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 13:08:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/630e4d296e21402e8a5f514c35e80a8e : SUCCESS in 1h 36m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7616a7df3fac4b0fbc32a8ceaeba0736 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6210b29772854524bdb470e0a02de7ea : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d75566f8955449ffa0df61635caf7615 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d021e98c176b4542bd56444bc8ab9586 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097c6d60e5924da1a2795cfcd6e05a74 : SUCCESS in 3m 35s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/2ebcaa90e8de42eca643461962c87f36 : SUCCESS in 43m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5a16dd5f5a1878a68a573c3231dbe3fb645581ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 13:08:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"06e7b8daaf24b9b44bd7e1fe60ea2ccf61a71706","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 13:10:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f6d054c9cc947b0bc8e6e42094267d9 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3e9862cccac949156001f02270b0fb4e43f14c5","revisions":{"b3e9862cccac949156001f02270b0fb4e43f14c5":{"kind":"REWORK","_number":1,"created":"2020-01-31 09:55:52.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/81/705181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/81/705181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/81/705181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/81/705181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/81/705181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/81/705181/1"}}},"commit":{"parents":[{"commit":"2dd526d1ed0f1f886fff0c2edfc2929dc6f58d1b","subject":"Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2dd526d1ed0f1f886fff0c2edfc2929dc6f58d1b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-31 09:55:10.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-31 09:55:10.000000000","tz":60},"subject":"Ignore releasenote artifacts files.","message":"Ignore releasenote artifacts files.\n\nChange-Id: I2a5a8b83d965ab6837ae01640ea4525284a0e7be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b3e9862cccac949156001f02270b0fb4e43f14c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b3e9862cccac949156001f02270b0fb4e43f14c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
