)]}'
{"id":"openstack%2Foslo.log~710791","triplet_id":"openstack%2Foslo.log~master~I0ad972f48ee6bc331b96aa67bd7e69c97fa8e2c3","project":"openstack/oslo.log","branch":"master","topic":"drop-six","hashtags":[],"change_id":"I0ad972f48ee6bc331b96aa67bd7e69c97fa8e2c3","subject":"drop use of six","status":"MERGED","created":"2020-03-02 14:54:24.000000000","updated":"2020-03-21 21:57:16.000000000","submitted":"2020-03-21 21:56:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710791-1584827768277-354c625b","meta_rev_id":"b44c25613d4f215c6e6092634116d26c462e9ea2","_number":710791,"virtual_id_number":710791,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-21 21:56:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-21 20:41:53.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-12 09:45:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-21 20:41:53.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-12 09:45:19.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-03-21 20:41:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-21 21:56:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"147cfc7ba50ae011ae3a70c52a7922d27efe9949","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-02 14:54:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0853c38d3f07cc18017f12da09f1bb96052bd89d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 16:14:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/37179621df97407bb77ad8429f1d05d3 : SUCCESS in 3m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63ed767645dd4b38bc406c939bef5505 : SUCCESS in 1h 17m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11d451c38dfc40dcb3cb252837de8aff : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/901b4cbaa20f4ea68be940d1aad82122 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bafdf097c244985b494514046667858 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f10a545c95345c1b35304ec51ca97de : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae73fd2a893a488cbd3fb14a0be1b22a : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe61e1313e734dc895df91d94a9e80db : SUCCESS in 5m 30s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/448e46d3ad6a4b6b8ac492824f3e9b71 : SUCCESS in 1h 04m 11s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/c0a9946c47734235a09cac3c8d60147e : SUCCESS in 44m 41s","accounts_in_message":[],"_revision_number":1},{"id":"64598c433d16133886fdfa80d2128e811e156145","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-11 14:42:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f823d15647acc8e45ef476a20f39520617962990","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-11 14:42:54.000000000","message":"Patch Set 1: Code-Review-1\n\nit\u0027s a nit but I think we should address that comment now","accounts_in_message":[],"_revision_number":1},{"id":"567ce2a6f8de7d158bab1bfc1f119bd23992d092","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-12 08:25:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45e85703672055d2aff10ac00b03aef80361712a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-12 08:28:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b9c3897d8df14c63390411073b4c0ff244142d99","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-12 09:45:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5f893dea047deae23276ec963ce0a580bdb0b828","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 09:52:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4b58acacb294477fbff8d28e0298e4af : SUCCESS in 2m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5b68dd9c403486ca69665e784a56dd8 : SUCCESS in 1h 18m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5a673be56c244699f8f0d2f7564aae8 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ad397a0a5694b4fbcc7cb7439d65469 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95788e0edbf74299a991124f25c627ad : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03793c557ada4b24a28e9d9996a490e7 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/981f93bafa9f4d6c921f5447336fad76 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e402bc488d34eb79c3950be8059d33e : SUCCESS in 3m 39s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/a1f24bf2d52c4c84952b68d7700d2e28 : SUCCESS in 1h 06m 33s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/3e9766cc09104a1c8a25fdceed18564a : SUCCESS in 53m 10s","accounts_in_message":[],"_revision_number":2},{"id":"e46d812507e436e44ebb2a579fe7ae43b405e8f2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-21 20:41:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2b851e13fb120511133aa59f4a83ecaf5a8c3424","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 20:42:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0fb22d488e08ecaf5c2c45937bdd1e249eccd3d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 21:56:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/348894e4c0af44b69ca68ee38a78c18b : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5176c8aa75cb46c9b127a616a3867081 : SUCCESS in 1h 13m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/adcf76fee81c4808845eedfcdf45a731 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5358e22ddfc74cd089d78165349cbf15 : SUCCESS in 2m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a5b10371ea64adb90473bedd28c7470 : SUCCESS in 2m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98c380ca30fd423fbdcd643734d21ccf : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dca0b1b83e47411284f3608f8290f119 : SUCCESS in 3m 02s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/a77b24f47b9a4be0b942468fa1b70e61 : SUCCESS in 37m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1a07f973d1afaa0433e4d607020b5a612ebed9af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 21:56:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d56b41ae7c0ff451365234f2c7315342e4ccfe5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 21:57:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dd070fc56b2423591035a242fe094ad : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"55ef517065b63e20534bfccd22585a52d1f0f9f1","revisions":{"c0a07b5d870d3b26acab351e4b03359d072aa56e":{"kind":"REWORK","_number":1,"created":"2020-03-02 14:54:24.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/91/710791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/91/710791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/710791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/710791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/710791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/91/710791/1"}}},"commit":{"parents":[{"commit":"26abff6f5499a256ce9221e5512b8499d3e46e6f","subject":"remove outdated header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/26abff6f5499a256ce9221e5512b8499d3e46e6f"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-02 14:54:14.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-02 14:54:14.000000000","tz":60},"subject":"drop use of six","message":"drop use of six\n\nChange-Id: I0ad972f48ee6bc331b96aa67bd7e69c97fa8e2c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/c0a07b5d870d3b26acab351e4b03359d072aa56e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/c0a07b5d870d3b26acab351e4b03359d072aa56e"}]},"branch":"refs/heads/master"},"55ef517065b63e20534bfccd22585a52d1f0f9f1":{"kind":"REWORK","_number":2,"created":"2020-03-12 08:28:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/91/710791/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/91/710791/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/710791/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/710791/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/710791/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/91/710791/2"}}},"commit":{"parents":[{"commit":"26abff6f5499a256ce9221e5512b8499d3e46e6f","subject":"remove outdated header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/26abff6f5499a256ce9221e5512b8499d3e46e6f"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-02 14:54:14.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-12 08:28:33.000000000","tz":60},"subject":"drop use of six","message":"drop use of six\n\nChange-Id: I0ad972f48ee6bc331b96aa67bd7e69c97fa8e2c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/55ef517065b63e20534bfccd22585a52d1f0f9f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/55ef517065b63e20534bfccd22585a52d1f0f9f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
