)]}'
{"id":"openstack%2Foslo.log~711561","triplet_id":"openstack%2Foslo.log~master~I515a6b03b810a7c8b2246279e380973a29c626f4","project":"openstack/oslo.log","branch":"master","topic":"victoria-wallaby","hashtags":[],"change_id":"I515a6b03b810a7c8b2246279e380973a29c626f4","subject":"Add Victoria and Wallaby releases to versionutils","status":"MERGED","created":"2020-03-05 22:12:25.000000000","updated":"2020-03-06 01:09:00.000000000","submitted":"2020-03-06 01:07:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711561-1583456848889-30632db2","meta_rev_id":"c7668f2bd22347cf99dd1e8320a2de81690e1b31","_number":711561,"virtual_id_number":711561,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-06 01:07:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-05 22:25:11.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-05 22:33:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-05 22:33:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-05 22:25:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-05 22:33:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-06 01:07:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19f946fa96824e9781afd3689de1facda85c765e","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-03-05 22:12:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"773385f6447a7bdc47805badef7913465815d463","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-05 22:25:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"46cc30f025c732590af6617e7e35cecefc12db02","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-05 22:33:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eda5e0f73f031d016906cf0c715fedc6aa7024bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 23:33:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85e7e5d123994223aec2eae5394558a7 : SUCCESS in 1h 19m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3640ccdbb0a147b7b5278d59a0d645a7 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1f7af95450a46dc99b6d4db38309447 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/542a7709fcb54bae900faaa2cb7b7607 : SUCCESS in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2977c281118044dfb17f73358c2ea2e2 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f9e7469991b477abe42a4c2657f3b5d : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaf25369200543e6bd4d397364058927 : SUCCESS in 3m 22s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/0df3b6a8a41e432ca680bb7c642d0906 : SUCCESS in 59m 53s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/86582099496642aabb2682d7b8b37e73 : SUCCESS in 39m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3bb4e5ac4b8c5fcc604b0d04347fc5420fb1f05a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 23:33:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ee992f31fabefbd1d9f3ae768f729ab90a53897","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 01:07:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4cc0bf600004d279b941a815bd6e758 : SUCCESS in 1h 32m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02d5d14d711d47b289c5aa2dd79c228e : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd342a24d784ea3891093ac339b8946 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/104a5e33cc8a4da2846e2967fd9563b8 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e16d9dd328d644e89ea49e5b1b00b9f2 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3f2012a575d4f7b92d23c4c36e731cc : SUCCESS in 4m 26s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/6df8da9d449f49c5acdb5d36b92c1e51 : SUCCESS in 42m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6d74803fd356975d4f1a916c9af0d7cb3aad44ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 01:07:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a74ebb2daf208e6a0f8205d7fbc91ddab433d3b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 01:09:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e7b3c7f16ff4afcbcf75572ec28c8e5 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"adef9b6ecbecedad9836e96a092c32cc8a17eb97","revisions":{"adef9b6ecbecedad9836e96a092c32cc8a17eb97":{"kind":"REWORK","_number":1,"created":"2020-03-05 22:12:25.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/61/711561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/61/711561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/711561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/711561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/711561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/61/711561/1"}}},"commit":{"parents":[{"commit":"26abff6f5499a256ce9221e5512b8499d3e46e6f","subject":"remove outdated header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/26abff6f5499a256ce9221e5512b8499d3e46e6f"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-03-05 22:11:51.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-03-05 22:11:51.000000000","tz":0},"subject":"Add Victoria and Wallaby releases to versionutils","message":"Add Victoria and Wallaby releases to versionutils\n\nNow that we have names for the next two releases we can get them\nadded here so we\u0027ll be ready when development starts.\n\nChange-Id: I515a6b03b810a7c8b2246279e380973a29c626f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/adef9b6ecbecedad9836e96a092c32cc8a17eb97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/adef9b6ecbecedad9836e96a092c32cc8a17eb97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
