)]}'
{"id":"openstack%2Foslo.log~718710","triplet_id":"openstack%2Foslo.log~master~I62a32ed46a400b62ead8c1e6e64acee658e2a769","project":"openstack/oslo.log","branch":"master","topic":"bug/1871840","hashtags":[],"change_id":"I62a32ed46a400b62ead8c1e6e64acee658e2a769","subject":"Add missing SYSLOG_FACILITY to JournalHandler","status":"MERGED","created":"2020-04-09 14:12:20.000000000","updated":"2020-06-01 21:58:58.000000000","submitted":"2020-05-28 11:00:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718710-1590663600407-06556da8","meta_rev_id":"5e1eb0360711ed5ab850f3185f2f2d4b7e8055e3","_number":718710,"virtual_id_number":718710,"owner":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 11:00:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-05-28 09:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-05-14 14:41:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},{"value":2,"date":"2020-05-25 14:14:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-05-28 09:43:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-14 14:41:37.000000000","updated_by":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"reviewer":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"state":"REVIEWER"},{"updated":"2020-05-25 14:14:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-28 09:43:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-05-28 11:00:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e4383fe5c6047e95d4f140ca44288b2f018adfe","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-04-09 14:12:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80418460832503c603fc0e866b48479eb5f1e807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 15:34:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/563c1e293c6447be9e0c57303d3ee276 : FAILURE in 23m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a51fad58abb45eb88378fd5da3bfc11 : FAILURE in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e591e81c3014747b5bd0403ade6b4c0 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/564a3c0eecf449c4a25d7e2323b7a137 : FAILURE in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c78a5dfcff974f74a9a67a18b3546fd4 : FAILURE in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52fe377555a14644bcabf07cf068e027 : FAILURE in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0c4ff64b6db4e709aa6a0a79a04bf3d : FAILURE in 5m 41s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/fabf8dbd84934104ba4f21180c761f05 : FAILURE in 1h 13m 42s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/55dd7fc32c41493898c1d9c797a25208 : FAILURE in 22m 37s","accounts_in_message":[],"_revision_number":1},{"id":"15fc7fd21fa35d56bd58a91c64f05e9be4a37442","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-01 16:04:49.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a711e4e3e36d6c13a4ef1a580d85841acf77f35d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-01 17:20:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94cf20776c394c8075fcd64b3115c06a209b6faf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-01 17:20:58.000000000","message":"Patch Set 2: Code-Review+2\n\nFixed the failing tests. This should be good now","accounts_in_message":[],"_revision_number":2},{"id":"a2cd08108de130fba772b04ffd28a79358461dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 18:07:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/351102a2f6084689919a90c93358e73e : POST_FAILURE in 18m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/875b3e2b485443fbbbd05f7ccaab3f58 : FAILURE in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0de870dcd854d849f11138aa070d9a2 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07d351b87b134c58b850f1e2e3e69f23 : FAILURE in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c41083447a0f488095801a51eb7f5404 : FAILURE in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b6b30c82bee4aaeabc5370922a00b1b : FAILURE in 4m 09s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/4aa6b62796094b2f8609e98c6d92eba6 : FAILURE in 43m 04s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/a7a0f884eb104cdf82822d945092f24d : POST_FAILURE in 17m 00s","accounts_in_message":[],"_revision_number":2},{"id":"078dfe017f105090b78e7f4a672c2dd60c55f744","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-05 11:25:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"94c82371d85788c5fbb774f2d2ccd898862ec09d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-05 12:17:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"56d4d4aa5942cdef78b33385501b11f64c1360f9","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-05 12:40:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5e5f03795cf4cb1afe03645a936f7699c8a5fe8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 12:47:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c75da4d358f429ab423e83edf59cc48 : SUCCESS in 1h 19m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c3da3916dd54cfbaf352cec65aaa0c2 : FAILURE in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2158d06584f4a9cb596b383cd09e57c : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95a013bb57fa4ddf82658e8fe281f137 : FAILURE in 9m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/265586a8659b4123a355ba46b559002c : FAILURE in 9m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5726bbfc03fb4818950ef8aa60c97702 : SUCCESS in 5m 13s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/daa1604117254ec4aed37795868cb765 : SUCCESS in 1h 09m 07s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/2f6c3f5f1f5f4652b0a806bb0f5d6d04 : SUCCESS in 40m 36s","accounts_in_message":[],"_revision_number":3},{"id":"5d35c52e2d964a71dcf57942202e32bee8565c60","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-06 09:38:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d582b87d5bccf494be14b382b6fd531755ee5119","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-06 09:56:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"84460cfb7093a8da999c0e0c0dc0af9f8a008b91","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-06 10:54:21.000000000","message":"Patch Set 4: Code-Review+2\n\nI suppose we want to backport these changes to stable branch to fix that there too on their, so it could be worth to append a related release note too (as a fix) to properly highlight this fix all along the road.","accounts_in_message":[],"_revision_number":4},{"id":"d8be6b4819047e36d002b951f819e480b95c89c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 11:15:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6263608036d148e68e384b4c8e56a877 : SUCCESS in 1h 31m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12f27a1a6e754a49a8a1061dcb1616f4 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bb0b63147584b8d8f66e4861d967a63 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c908a8bc2a048f98920a9b945334eda : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7f304d4d2c648e5a1eff86155e1f6eb : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc9d45d7c0ae41a2a2c34a31de0f3392 : SUCCESS in 4m 36s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/0bd4da24930f42d4ba24203dc9bd7356 : SUCCESS in 1h 23m 33s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/96b25ee8085e44b89bcffd9b53c6b00d : SUCCESS in 37m 58s","accounts_in_message":[],"_revision_number":4},{"id":"173fd11a1ff06db5d4b866f50a6f2a67375bed9d","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-07 12:06:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"62bdc7b947dcec89b577a94fed02014f6c6db7ab","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-07 12:47:44.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"de39c1abfa59721e715972c989ad3630cc277154","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:21:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c52edbfcdcc34647ae493a8c7b37c792 : SUCCESS in 1h 31m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b161166b9318413aa6764ead07281168 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad0423edecf45d195fc0f91fe491360 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e06c10a1092a4ce3b25de2c147417a48 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3c23d5ed30e4f22848d00c3771b0f2d : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df845d67dd9147e8ab7bf484d2a5dd92 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a09601005244a94842ba2ffc7fcaac8 : SUCCESS in 5m 19s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/37ecdabfe389481ba063530566535924 : SUCCESS in 59m 10s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/e93529b37bcd4ae8a28876745b351169 : SUCCESS in 45m 42s","accounts_in_message":[],"_revision_number":6},{"id":"d92d9b54d157b055392e15ed1e8c83886e7c726f","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-07 14:26:27.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 82751ce2a2b913a5d107133a15960c9c6d8fd594","accounts_in_message":[],"_revision_number":6},{"id":"02c958c7bf6dddb519d4556dbf15af63d1094959","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-07 14:26:40.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 911a45401c1292b1946c200aa0f6a98b9e246fd8","accounts_in_message":[],"_revision_number":6},{"id":"351c01099d0c6f09cb61f550e1df97c2dad662c9","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-07 14:27:17.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit b97e81dad2eee2d334c3b5fe74c499e2493b9cda","accounts_in_message":[],"_revision_number":6},{"id":"c68fc27e96a09a1a08df38c1bbc3f704710d7927","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2020-05-14 14:18:41.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"882a4645db33a1350502e0d6cadc1811104b9abe","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2020-05-14 14:41:19.000000000","message":"Removed Code-Review+1 by Olaf Seibert \u003co.seibert@syseleven.de\u003e\n","accounts_in_message":[],"_revision_number":6},{"id":"91128ba76c996fccf75bc55c692ca594f386aeb0","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2020-05-14 14:41:37.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f31e212410f11cc18f4f47f38c908a3fcc9ce1db","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-25 14:14:02.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"e7264a64b82d7e4fb9d9665c07cde65fadba4b9e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-28 09:43:15.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"11a8724518dfffccbaeb74fea0a490c1b85ffcbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:43:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ab18575b1f509c03d65788a4f9cc3713b3aee6cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 11:00:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"de30942dcf8c542f55db8d6cd0b0694fc69df710","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 11:00:00.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc408f4a0ff24a5ea3d16014d4ed9660 : SUCCESS in 1h 15m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/818a6ab8eaf9499c958fd5259b723ee1 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db5bb4e908dd4a489850659ce549e956 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e44d04ee9859423885d1c2a77196c9ce : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20357077172f47a99dc0809e1fd47ef9 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10536e6a3bfd433a96a0e7c7dbc1ad96 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b5cc99752c54a5bbf4769f5140d97fc : SUCCESS in 3m 33s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/1e97b361df684f3db895bcef7a574a75 : SUCCESS in 36m 40s","accounts_in_message":[],"_revision_number":6},{"id":"d6173c52afac4f2f4dbe2c90649ad7e80d522346","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 11:02:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75e885601bd34dd0a99810bf9417b12f : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90a4c72b38a14d02aa94bd3e3e17bba1 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":6},{"id":"e2ec98ee20e3653d093ab37d2ab31d5b562be570","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-01 21:58:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5b12e66f6283962d96884e4aa3fa7db7d4e062fa","revisions":{"121b6196e5ba02137b8790ffd6560958aa5c4b40":{"kind":"REWORK","_number":1,"created":"2020-04-09 14:12:20.000000000","uploader":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"ref":"refs/changes/10/718710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/10/718710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/1"}}},"commit":{"parents":[{"commit":"2aaf7b08a651a3360755f6401b78dcb77a13bfb9","subject":"Use unittest.mock instead of third party mock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2aaf7b08a651a3360755f6401b78dcb77a13bfb9"}]}],"author":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-04-09 13:02:20.000000000","tz":120},"committer":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-04-09 13:02:20.000000000","tz":120},"subject":"Add missing SYSLOG_FACILITY to JournalHandler","message":"Add missing SYSLOG_FACILITY to JournalHandler\n\nWithout SYSLOG_FACILITY log messages from journald forwards\nthe message to kern.log. This seems to be default behaviour as\nit treats the message as kernel message.\n\nThis option provides a proper facility to the log message\nand kern.log will not be filled up anymore.\n\nChange-Id: I62a32ed46a400b62ead8c1e6e64acee658e2a769\nCloses-Bug: #1871840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/121b6196e5ba02137b8790ffd6560958aa5c4b40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/121b6196e5ba02137b8790ffd6560958aa5c4b40"}]},"branch":"refs/heads/master"},"f301dec70d2fa692b47999f2628bf1372b8eed5b":{"kind":"REWORK","_number":2,"created":"2020-05-01 17:20:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/10/718710/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/10/718710/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/2"}}},"commit":{"parents":[{"commit":"1c1fd4e56cc29001837a4ccae3eae990c4bb3b82","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1c1fd4e56cc29001837a4ccae3eae990c4bb3b82"}]}],"author":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-04-09 13:02:20.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-05-01 17:20:06.000000000","tz":60},"subject":"Add missing SYSLOG_FACILITY to JournalHandler","message":"Add missing SYSLOG_FACILITY to JournalHandler\n\nWithout SYSLOG_FACILITY log messages from journald forwards the message\nto kern.log. This seems to be default behaviour as it treats the message\nas kernel message.\n\nThis option provides a proper facility to the log message and kern.log\nwill not be filled up anymore.\n\nChange-Id: I62a32ed46a400b62ead8c1e6e64acee658e2a769\nCloses-Bug: #1871840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f301dec70d2fa692b47999f2628bf1372b8eed5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f301dec70d2fa692b47999f2628bf1372b8eed5b"}]},"branch":"refs/heads/master"},"cf0d75dbf7f7154bc1d55ff60c8d5818bbcdd8af":{"kind":"REWORK","_number":3,"created":"2020-05-05 11:25:26.000000000","uploader":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"ref":"refs/changes/10/718710/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/10/718710/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/3"}}},"commit":{"parents":[{"commit":"1c1fd4e56cc29001837a4ccae3eae990c4bb3b82","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1c1fd4e56cc29001837a4ccae3eae990c4bb3b82"}]}],"author":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-04-09 13:02:20.000000000","tz":120},"committer":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-05-05 11:23:11.000000000","tz":120},"subject":"Add missing SYSLOG_FACILITY to JournalHandler","message":"Add missing SYSLOG_FACILITY to JournalHandler\n\nWithout SYSLOG_FACILITY log messages from journald forwards the message\nto kern.log. This seems to be default behaviour as it treats the message\nas kernel message.\n\nThis option provides a proper facility to the log message and kern.log\nwill not be filled up anymore.\n\nChange-Id: I62a32ed46a400b62ead8c1e6e64acee658e2a769\nCloses-Bug: #1871840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/cf0d75dbf7f7154bc1d55ff60c8d5818bbcdd8af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/cf0d75dbf7f7154bc1d55ff60c8d5818bbcdd8af"}]},"branch":"refs/heads/master"},"61f513904284471565041caf48239ceeaa825dde":{"kind":"REWORK","_number":4,"created":"2020-05-06 09:38:50.000000000","uploader":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"ref":"refs/changes/10/718710/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/10/718710/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/4"}}},"commit":{"parents":[{"commit":"1c1fd4e56cc29001837a4ccae3eae990c4bb3b82","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1c1fd4e56cc29001837a4ccae3eae990c4bb3b82"}]}],"author":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-04-09 13:02:20.000000000","tz":120},"committer":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-05-06 09:38:30.000000000","tz":120},"subject":"Add missing SYSLOG_FACILITY to JournalHandler","message":"Add missing SYSLOG_FACILITY to JournalHandler\n\nWithout SYSLOG_FACILITY log messages from journald forwards the message\nto kern.log. This seems to be default behaviour as it treats the message\nas kernel message.\n\nThis option provides a proper facility to the log message and kern.log\nwill not be filled up anymore.\n\nChange-Id: I62a32ed46a400b62ead8c1e6e64acee658e2a769\nCloses-Bug: #1871840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/61f513904284471565041caf48239ceeaa825dde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/61f513904284471565041caf48239ceeaa825dde"}]},"branch":"refs/heads/master"},"8c732084f86e95870401b6bfb5d59b38020f60af":{"kind":"REWORK","_number":5,"created":"2020-05-07 12:06:14.000000000","uploader":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"ref":"refs/changes/10/718710/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/10/718710/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/5"}}},"commit":{"parents":[{"commit":"1c1fd4e56cc29001837a4ccae3eae990c4bb3b82","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1c1fd4e56cc29001837a4ccae3eae990c4bb3b82"}]}],"author":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-04-09 13:02:20.000000000","tz":120},"committer":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-05-07 12:03:52.000000000","tz":120},"subject":"Add missing SYSLOG_FACILITY to JournalHandler","message":"Add missing SYSLOG_FACILITY to JournalHandler\n\nWithout SYSLOG_FACILITY log messages from journald forwards the message\nto kern.log. This seems to be default behaviour as it treats the message\nas kernel message.\n\nThis option provides a proper facility to the log message and kern.log\nwill not be filled up anymore.\n\nChange-Id: I62a32ed46a400b62ead8c1e6e64acee658e2a769\nCloses-Bug: #1871840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/8c732084f86e95870401b6bfb5d59b38020f60af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/8c732084f86e95870401b6bfb5d59b38020f60af"}]},"branch":"refs/heads/master"},"5b12e66f6283962d96884e4aa3fa7db7d4e062fa":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-05-07 12:47:44.000000000","uploader":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"ref":"refs/changes/10/718710/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/10/718710/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/10/718710/6"}}},"commit":{"parents":[{"commit":"105eadbbac84290818e5809c58bd44f43bea3d2d","subject":"Remove monotonic usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/105eadbbac84290818e5809c58bd44f43bea3d2d"}]}],"author":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-04-09 13:02:20.000000000","tz":120},"committer":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-05-07 12:46:08.000000000","tz":120},"subject":"Add missing SYSLOG_FACILITY to JournalHandler","message":"Add missing SYSLOG_FACILITY to JournalHandler\n\nWithout SYSLOG_FACILITY log messages from journald forwards the message\nto kern.log. This seems to be default behaviour as it treats the message\nas kernel message.\n\nThis option provides a proper facility to the log message and kern.log\nwill not be filled up anymore.\n\nChange-Id: I62a32ed46a400b62ead8c1e6e64acee658e2a769\nCloses-Bug: #1871840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/5b12e66f6283962d96884e4aa3fa7db7d4e062fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/5b12e66f6283962d96884e4aa3fa7db7d4e062fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
