)]}'
{"id":"openstack%2Foslo.log~725422","triplet_id":"openstack%2Foslo.log~master~Id250b92d11aeb6fa02e264610bfe8e8c71df7e26","project":"openstack/oslo.log","branch":"master","topic":"monotonic","hashtags":[],"change_id":"Id250b92d11aeb6fa02e264610bfe8e8c71df7e26","subject":"Remove monotonic usage","status":"MERGED","created":"2020-05-04 22:26:35.000000000","updated":"2020-05-05 14:24:50.000000000","submitted":"2020-05-05 14:23:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725422-1588688585753-111d3b5a","meta_rev_id":"d7fc3973945a18561fe82bd959db59b6a3b41ce9","_number":725422,"virtual_id_number":725422,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-05 14:23:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-05-05 12:45:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-05-05 10:06:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-05-05 12:45:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-05 10:06:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-05 12:45:02.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-05-05 14:23:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21efc77a1ca0594d2d6b61976a9b1eea9ce4e9df","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-04 22:26:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66184dfe148953746b1d7066dd9bb0657d7e6ef8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 00:01:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/22a4c121ff9348479c8191d65b9592a1 : SUCCESS in 3m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2df3cd13a3404f159a2b5932cdc28966 : SUCCESS in 1h 32m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a21b786726449fb814fd2aa97d5e4d1 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a310c54be9e54ca09b4b3a9cf249764b : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a369dc651be441d812787bb5f31ac5b : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aea8c7c60f55452bb5d038946b5917ce : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ad0531b8007487385a7445b99a252f1 : SUCCESS in 5m 00s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/f25c5497625b455c8e1e5e2c483af773 : SUCCESS in 1h 10m 02s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/c8c0b6a273984a4d98ef2168585cd420 : SUCCESS in 54m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d4324c319e8a9a5429b7addcceadb386fa1bf352","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-05 10:06:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b367cb238104b2332b632546e12c56af40a38e7e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-05 12:45:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3859659b51f50bb017e12e1cfeb763561c375e7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 12:45:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f43571672c5ff43f439dedebbd75f875762228d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 14:23:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1a3b94cccc5e4d3095a9adf48729e3ef : SUCCESS in 2m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/090c50a81c0d46aaabc1d020e8b77a34 : SUCCESS in 1h 35m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/499829318a114e309616de99d797ffab : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/472f6e75c220404ba2da7d1f92e693ed : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb575b5447ec48048bb43dbbc761670f : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a984bcdd90e0492bafd87d8d86384cf6 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f21b5686f90e41fca7975272893219f5 : SUCCESS in 4m 38s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/077229bb88714d208a93aba40accb6b8 : SUCCESS in 45m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3b6f2442917e15205645bd81f3b5dbce0de8c963","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 14:23:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"99661b36d5741baf22b70de8435078849f0583fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 14:24:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a931f3508c4b4a71a9bf2b85e1459f57 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"105eadbbac84290818e5809c58bd44f43bea3d2d","revisions":{"105eadbbac84290818e5809c58bd44f43bea3d2d":{"kind":"REWORK","_number":1,"created":"2020-05-04 22:26:35.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/22/725422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/22/725422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/22/725422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/22/725422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/22/725422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/22/725422/1"}}},"commit":{"parents":[{"commit":"1c1fd4e56cc29001837a4ccae3eae990c4bb3b82","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1c1fd4e56cc29001837a4ccae3eae990c4bb3b82"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-04 22:26:30.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-04 22:26:30.000000000","tz":-300},"subject":"Remove monotonic usage","message":"Remove monotonic usage\n\nThe monotonic package was needed for monotonic time operations when\nrunning under Python runtimes older than 3.3. Since we now only support\nversions higher than this, this third party package requirement can now\nbe removed.\n\nChange-Id: Id250b92d11aeb6fa02e264610bfe8e8c71df7e26\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/105eadbbac84290818e5809c58bd44f43bea3d2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/105eadbbac84290818e5809c58bd44f43bea3d2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
