)]}'
{"id":"openstack%2Foslo.log~726156","triplet_id":"openstack%2Foslo.log~stable%2Fussuri~I62a32ed46a400b62ead8c1e6e64acee658e2a769","project":"openstack/oslo.log","branch":"stable/ussuri","topic":"bug/1871840","hashtags":[],"change_id":"I62a32ed46a400b62ead8c1e6e64acee658e2a769","subject":"Add missing SYSLOG_FACILITY to JournalHandler","status":"MERGED","created":"2020-05-07 14:26:27.000000000","updated":"2020-05-28 16:15:32.000000000","submitted":"2020-05-28 16:11:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726156-1590682303077-c6985dc5","meta_rev_id":"d37933622551d63d43645709560a307fb8f05cf5","_number":726156,"virtual_id_number":726156,"owner":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 16:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-25 18:57:41.000000000","_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-05-28 09:43:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-05-27 11:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-05-28 14:47:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 11:42:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-28 14:47:59.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-05-28 16:11:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9d0676baf8afafb87ae1da8301120184af82061","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-07 14:26:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"71fdba2d5e86281b0cad27318768703eec3e131d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 16:06:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c7b0961fcf54f3b960ddc4a6d7c5ff6 : SUCCESS in 1h 33m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb4fe64463ee4db5b71e2de85e66f7be : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c22637d006a94bd6946885bf0e8e07bd : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0da6d66f9714874bfa017bb2b7fad00 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44f7e56f453349eab61dba2b05b22721 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b8f22ec7a7e46a28fcb3c7313602db0 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b3cd0c239ac4347833146ced563a801 : SUCCESS in 8m 14s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/0aa83a32eb4c453fac74177f5199d3c3 : SUCCESS in 1h 27m 23s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/87cc057894974b0296f0b57d74032350 : SUCCESS in 37m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2b95e43bc17132578790e48ec7a7841de7f91959","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-25 12:58:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7efb745d44d3ee5ca39fdf8385442dd1d8ad0352","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-25 15:14:42.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4b4fca9fa81e1806ab094b5584bd6092199996a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 17:22:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f89e363cd51049fb8150b6c982f658e7 : TIMED_OUT in 2h 06m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78a8176587e3423f924a3d2b325604b5 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/997292afc53d43b28f63230a05882436 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9108c8a05c7c4f34999b4f98fd9abfba : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14bcccb5d49543c2afea4d30618c33ff : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce78f5ff0764f67b205a136555cdf4e : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04a2503d32f14192aa1d0e93c409b751 : SUCCESS in 3m 41s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/f7239793853b47d1949768cebaef9fbf : SUCCESS in 1h 05m 28s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/b51a8c2ca9f244848992a2762091a582 : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6cdfcff5544266a51580be1a4671a4a86de4e49c","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-25 18:57:41.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a4ecce72a659e562e2192c00ca51713f7d08018f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-27 11:41:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"05ee8e2b26a57614d69973d18292df7986afd996","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-27 11:42:33.000000000","message":"Patch Set 2: Code-Review+2\n\nClean backport LGTM","accounts_in_message":[],"_revision_number":2},{"id":"6fdacbe440e58e7e10261f1bfdd95506c562c444","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 13:01:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f6727b7f3dd4387a1d2794bcbf6f173 : SUCCESS in 1h 18m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/775f73879a4e406f99934ff040cab986 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/310cec7607474493ae2dec0df99d7807 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/692dc6de543b4be3b4f20babbc5e0998 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98370cade34c402daed464869e364796 : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29dc0fd2afc14f57a0069ef6fb460402 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77bbeab96fee4dacbf5cd15bd0e54db5 : SUCCESS in 5m 28s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/f21d68a40cb74c3b8573d2d8c43178bc : SUCCESS in 1h 11m 02s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/7bee23800f45475f9a426eb2bba489d9 : SUCCESS in 44m 22s","accounts_in_message":[],"_revision_number":2},{"id":"02ce54f725ebfba0dcaad83d5670b1788f55a0f6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-28 09:43:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5e78e70331ab0fb8b90e38f43e4639e516dfc789","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-28 09:44:01.000000000","message":"Patch Set 2:\n\nWill +W once master lands","accounts_in_message":[],"_revision_number":2},{"id":"274ecc9ee824c8887a804cdaf108afbc2516a539","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-28 14:47:59.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dc84498d559719af6de4aa5b1cc079a6905659bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 14:48:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a83501bf4ab4a453f10ec4d0b9a51f629def514","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-28 14:49:02.000000000","message":"Topic changed from bug/1871840-stable/ussuri to bug/1871840","accounts_in_message":[],"_revision_number":2},{"id":"a73642b12ebc327ed4f0df6d975f9776d2d744bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 16:11:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54adb4c0e55242b89c53838fcdd5dd70 : SUCCESS in 1h 21m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aefebd767ec14f39b776ef7bb6d6af36 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b6e045ec3094531b40768f1cb67c6ec : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec1256ea1b764d75ae612b0d2e1ea4cf : SUCCESS in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce5910b35a114d5aaf786a78cf6c1498 : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843e91960d6b40c9b1a468569fb98c8e : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb0e80d2c963440986b2abe8033392ce : SUCCESS in 5m 35s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/00929bc03c1e48c4befa645609181bde : SUCCESS in 38m 45s","accounts_in_message":[],"_revision_number":2},{"id":"95884b691e34fb3214ffdc0cee1b5f9b77296c18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 16:11:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d37933622551d63d43645709560a307fb8f05cf5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 16:15:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/359fbc41f01e45998d01d666480fd646 : SUCCESS in 1m 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a596d1cc0f9249f690175bd72c95b73a : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a67f7b5ce5b956918640784d9a180873676d5820","revisions":{"82751ce2a2b913a5d107133a15960c9c6d8fd594":{"kind":"REWORK","_number":1,"created":"2020-05-07 14:26:27.000000000","uploader":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"ref":"refs/changes/56/726156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/56/726156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/56/726156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/56/726156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/56/726156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/56/726156/1"}}},"commit":{"parents":[{"commit":"e1930c9e7a371c9af30735d2a1d427d88a1c0b55","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e1930c9e7a371c9af30735d2a1d427d88a1c0b55"}]}],"author":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-04-09 13:02:20.000000000","tz":120},"committer":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-05-07 14:26:27.000000000","tz":0},"subject":"Add missing SYSLOG_FACILITY to JournalHandler","message":"Add missing SYSLOG_FACILITY to JournalHandler\n\nWithout SYSLOG_FACILITY log messages from journald forwards the message\nto kern.log. This seems to be default behaviour as it treats the message\nas kernel message.\n\nThis option provides a proper facility to the log message and kern.log\nwill not be filled up anymore.\n\nChange-Id: I62a32ed46a400b62ead8c1e6e64acee658e2a769\nCloses-Bug: #1871840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/82751ce2a2b913a5d107133a15960c9c6d8fd594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/82751ce2a2b913a5d107133a15960c9c6d8fd594"}]},"branch":"refs/heads/stable/ussuri"},"a67f7b5ce5b956918640784d9a180873676d5820":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-25 15:14:42.000000000","uploader":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"ref":"refs/changes/56/726156/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/56/726156/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/56/726156/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/56/726156/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/56/726156/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/56/726156/2"}}},"commit":{"parents":[{"commit":"e1930c9e7a371c9af30735d2a1d427d88a1c0b55","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e1930c9e7a371c9af30735d2a1d427d88a1c0b55"}]}],"author":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-04-09 13:02:20.000000000","tz":120},"committer":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-05-25 15:13:23.000000000","tz":120},"subject":"Add missing SYSLOG_FACILITY to JournalHandler","message":"Add missing SYSLOG_FACILITY to JournalHandler\n\nWithout SYSLOG_FACILITY log messages from journald forwards the message\nto kern.log. This seems to be default behaviour as it treats the message\nas kernel message.\n\nThis option provides a proper facility to the log message and kern.log\nwill not be filled up anymore.\n\nCherry-picked from commit 5b12e66f6283962d96884e4aa3fa7db7d4e062fa\nChange-Id: I62a32ed46a400b62ead8c1e6e64acee658e2a769\nCloses-Bug: #1871840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a67f7b5ce5b956918640784d9a180873676d5820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a67f7b5ce5b956918640784d9a180873676d5820"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
