)]}'
{"id":"openstack%2Foslo.log~726157","triplet_id":"openstack%2Foslo.log~stable%2Ftrain~I62a32ed46a400b62ead8c1e6e64acee658e2a769","project":"openstack/oslo.log","branch":"stable/train","topic":"bug/1871840","hashtags":[],"change_id":"I62a32ed46a400b62ead8c1e6e64acee658e2a769","subject":"Add missing SYSLOG_FACILITY to JournalHandler","status":"MERGED","created":"2020-05-07 14:26:40.000000000","updated":"2020-05-28 16:30:04.000000000","submitted":"2020-05-28 16:27:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726157-1590683257403-b12ba7e5","meta_rev_id":"f2c9b3e0892254c9d2fab113aba9adeff1539692","_number":726157,"virtual_id_number":726157,"owner":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 16:27:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-05-28 09:44:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-05-27 11:39:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-05-28 14:48:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 11:39:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-28 14:48:07.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-05-28 16:27:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"331f100dbb00a78725e55f6833e6c18163b17002","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-07 14:26:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"39fa45160c2fd36529e103a7f524baf74a80247c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 16:27:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/da9c833017a547af8a15b2ae53db2d82 : FAILURE in 1h 49m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/407254bfa55745d2bf1ad2e064084b1c : SUCCESS in 1h 35m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0cd695c6f744bb3a3655a6b0a0293ad : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b224afbef1f4178a3286ed4f665308a : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42eee8771ef84cfcae61a743b45437d4 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f097caf470c1487586543e7808f4d355 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63701c50b59a450fb177d35f056122b0 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fe5ea1b1e9e4e74af247c8ee0b2bee4 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0e3659fb24c4ff18e0262e48ab52329 : SUCCESS in 4m 55s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/0ab6d7e6bb6c43028f67a8698db6899b : SUCCESS in 56m 43s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/0e6c73db51f645c39c5dbf3a71a8ae6a : SUCCESS in 43m 33s","accounts_in_message":[],"_revision_number":1},{"id":"8d9ba16ba25dfe4d73178f3695380063af40ec5f","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-07 21:53:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"755c572be7206669e67be44eed3bf120fae3777a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 00:06:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d9022eb87b474fc5a42775f441a0e52e : TIMED_OUT in 2h 10m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07113bf6c2624262888d6672268e5330 : TIMED_OUT in 2h 10m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c419408b2a0f4297b4accbcf69518256 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae0f6b8b92024cea9803fc186ebbd1ed : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d181091e284d4fdf85a2d29b8f9a488f : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/328e68ca2c314c3ea0b4c4ff55e19654 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f54ca4c744ed4936aba6c9c64a466b30 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da89bcbf32044633a413e97707511f50 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4605c8168f44839b1056472878377bb : SUCCESS in 5m 04s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/779bbcf8775547ee91fd5d806162bc37 : SUCCESS in 1h 13m 47s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/38623655bf4c4066abeae43173f6d57f : SUCCESS in 36m 53s","accounts_in_message":[],"_revision_number":1},{"id":"53798dec50ff415a84def1902914192770a03156","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-25 13:03:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"389b90fd779f041d782bad17bc7bafccd1c76ce2","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-25 15:17:11.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7683102ebe9c60dfd6c43c5b8cb933a4572e59a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 16:52:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3287eb447bc84bff88de69f72d84fde7 : SUCCESS in 1h 31m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c6fd56c744d48659d17be5d841d35f5 : SUCCESS in 1h 23m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89a889e82e2c4289832674e62588a1c0 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02c91b9f0c8547dfaf7bc91cc88b497f : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a93116c25c5a4437bb0a186c28997c27 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbff72ceb6d04ec69dd72a6194178e5c : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c63ad520a194a63916d771534cc7cfb : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e3694ec1154752b86c56d08f37db3f : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c31add383f949c6875a048e5a3ac57c : SUCCESS in 3m 21s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/fcdc1f517a0141eb9e6b8ad97c9b921b : SUCCESS in 56m 29s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/356edde817d74ae986ee2f8964a02f33 : SUCCESS in 43m 27s","accounts_in_message":[],"_revision_number":2},{"id":"a5f40a8aa0194e0629dd9218918628eafe090de9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-27 11:39:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ab2ad771fb49d6d7e690a55219eb3e5b3abc468e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-28 09:44:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0f57a3cba22081caebd39bd9a8069f7c02cff3cd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-28 14:48:07.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e9d28c5a41a64bd0398317772c81f478c95f22f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 14:48:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5b5a3bede25270ccb198d7ac67f3d766702eb89a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-28 14:48:52.000000000","message":"Topic changed from bug/1871840-stable/train to bug/1871840","accounts_in_message":[],"_revision_number":2},{"id":"947763bd70d3982d5b72c82fb9abfcef72c5ed71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 16:27:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f16db3937810bd20bb00d6729b4d712290abf6bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 16:27:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/87cea7fc60e545d9bfa6d55834ba6aaa : SUCCESS in 1h 37m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ac754c282904aba83426dc704e5ced7 : SUCCESS in 1h 21m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9dbb64465c904233806cb849eda189fa : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e96e368526424df382b21c4635b3625c : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2813e3f2113e4c8c8401e28ae820b11b : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34b3e762ad434de2ab2842b6fbc0f2c4 : SUCCESS in 8m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f13b0376cb634eebb22d0c3a5b2f3530 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0239de04e02b4597a2f6367ed422e58a : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/126e74f0e6a3405483fa8cfc9ca41081 : SUCCESS in 7m 11s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/4390d710b96343d6a9c831eb1979e5ef : SUCCESS in 42m 42s","accounts_in_message":[],"_revision_number":2},{"id":"f2c9b3e0892254c9d2fab113aba9adeff1539692","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 16:30:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c3253094684e09ab2df51a1f3b0a67 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cdb66c470ce45ca92061e67048c4a6f : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"65ee5f5bbb3ec759bfa1bf1dbccb02c3b348bad0","revisions":{"911a45401c1292b1946c200aa0f6a98b9e246fd8":{"kind":"REWORK","_number":1,"created":"2020-05-07 14:26:40.000000000","uploader":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"ref":"refs/changes/57/726157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/57/726157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/57/726157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/57/726157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/57/726157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/57/726157/1"}}},"commit":{"parents":[{"commit":"2f8169185f99c87a22d9af5e45075b928596987d","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2f8169185f99c87a22d9af5e45075b928596987d"}]}],"author":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-04-09 13:02:20.000000000","tz":120},"committer":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-05-07 14:26:40.000000000","tz":0},"subject":"Add missing SYSLOG_FACILITY to JournalHandler","message":"Add missing SYSLOG_FACILITY to JournalHandler\n\nWithout SYSLOG_FACILITY log messages from journald forwards the message\nto kern.log. This seems to be default behaviour as it treats the message\nas kernel message.\n\nThis option provides a proper facility to the log message and kern.log\nwill not be filled up anymore.\n\nChange-Id: I62a32ed46a400b62ead8c1e6e64acee658e2a769\nCloses-Bug: #1871840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/911a45401c1292b1946c200aa0f6a98b9e246fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/911a45401c1292b1946c200aa0f6a98b9e246fd8"}]},"branch":"refs/heads/stable/train"},"65ee5f5bbb3ec759bfa1bf1dbccb02c3b348bad0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-25 15:17:11.000000000","uploader":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"ref":"refs/changes/57/726157/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/57/726157/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/57/726157/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/57/726157/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/57/726157/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/57/726157/2"}}},"commit":{"parents":[{"commit":"2f8169185f99c87a22d9af5e45075b928596987d","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2f8169185f99c87a22d9af5e45075b928596987d"}]}],"author":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-04-09 13:02:20.000000000","tz":120},"committer":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-05-25 15:16:56.000000000","tz":120},"subject":"Add missing SYSLOG_FACILITY to JournalHandler","message":"Add missing SYSLOG_FACILITY to JournalHandler\n\nWithout SYSLOG_FACILITY log messages from journald forwards the message\nto kern.log. This seems to be default behaviour as it treats the message\nas kernel message.\n\nThis option provides a proper facility to the log message and kern.log\nwill not be filled up anymore.\n\nCherry-picked from commit 5b12e66f6283962d96884e4aa3fa7db7d4e062fa\nCherry-picked from commit 82751ce2a2b913a5d107133a15960c9c6d8fd594\nChange-Id: I62a32ed46a400b62ead8c1e6e64acee658e2a769\nCloses-Bug: #1871840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/65ee5f5bbb3ec759bfa1bf1dbccb02c3b348bad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/65ee5f5bbb3ec759bfa1bf1dbccb02c3b348bad0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
