)]}'
{"id":"openstack%2Foslo.log~726159","triplet_id":"openstack%2Foslo.log~stable%2Fstein~I62a32ed46a400b62ead8c1e6e64acee658e2a769","project":"openstack/oslo.log","branch":"stable/stein","topic":"bug/1871840","hashtags":[],"change_id":"I62a32ed46a400b62ead8c1e6e64acee658e2a769","subject":"Add missing SYSLOG_FACILITY to JournalHandler","status":"MERGED","created":"2020-05-07 14:27:17.000000000","updated":"2020-05-28 16:29:54.000000000","submitted":"2020-05-28 16:27:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726159-1590683258135-69ea3769","meta_rev_id":"2fb800d3f597295dbbfc103ddb3d0bb6c9ae8823","_number":726159,"virtual_id_number":726159,"owner":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 16:27:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-05-28 14:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-05-27 11:39:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-05-28 14:48:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 11:39:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-28 14:48:21.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-05-28 16:27:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37cfe25d5af781a2a42833a5f0bc1b2a15664586","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-07 14:27:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7662f36ca7f8d7ad37343f7e90f3caa5c7afdda7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 16:33:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e5662ec1b11c4a8e89bbf5bb6d51d5a6 : FAILURE in 1h 36m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1949e57b90c44ed9be66e96968b2dcc2 : SUCCESS in 1h 19m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5bcb31aca51416b8ee993ede145fa3e : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd6fcd5d44f549f29851bca39be5b8ad : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3042cc4545e740969c575848b22e4e25 : SUCCESS in 7m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1251ddbb57e24441a8c52a526f26daeb : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5e449a0a29944aba5ea6434eedcb86d : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35b48b85b474bc88aa0e28ec175a1cb : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8fe746514234fe18646422b777b6184 : SUCCESS in 5m 38s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/2516c8d4b65349a59ca7838c76f12e38 : SUCCESS in 54m 52s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/9141b1f18464477aa73d382779d9efe0 : SUCCESS in 42m 42s","accounts_in_message":[],"_revision_number":1},{"id":"5d1331546807671219794503e0776e86a53611bd","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-07 21:53:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"79b0065b4b5558b6a1ba61dcde962e372011f0a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 23:40:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/83b3221492f0488cadd11d6ceccbb78b : SUCCESS in 1h 44m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/054fbd1cc0d54cb89d1385c14238d926 : SUCCESS in 1h 20m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba852557656148dba4bd29228bd835d9 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56df261a83cd45e89d240877083583c7 : SUCCESS in 6m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b984865549ba45298b70ea8deaa1a06e : SUCCESS in 4m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1e243487f99e45b18ddb140c73ac1594 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4999f6932b50444ab87163744dbc29b0 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5c44096dc2d43e69a3473abb3ca4c9b : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fa7627b246840bcb7995d979c64c8b7 : SUCCESS in 4m 09s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/531121036461438abf369abb17f5dd19 : SUCCESS in 59m 18s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/8dbda786db024a91b818d34e70cef40d : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7d609e9e1134a7c9c8902cc150a3e89e4f1973b9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-25 13:05:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b72ed13c4e353c8859f5308a2878a36570ff6523","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-25 15:27:01.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d045b37a1cc5659bf847565a85717c75127414ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 16:57:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6fa8b332a92b4b5f8f1c87a397556e8e : SUCCESS in 1h 25m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3754a3158cc47b99a4caadc4869945f : SUCCESS in 1h 13m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b6c140ff72a4d00847b027ac59dd1a9 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a33947cd6e1c45a0b81ac6645dd950ac : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2242f88c994c446ab71c37de214f51b7 : SUCCESS in 3m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cd9a8f45ec214e0ba252ee5f220b2d41 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aab14a5965b94567b9670b4b9f0542be : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5a76fc03c964a318201a18e006b7d3c : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14e5793d46fc45b68d906fccd27be258 : SUCCESS in 3m 31s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/2228506b11414c08ba34084ad0d89e75 : SUCCESS in 1h 05m 12s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/b93faa39717b419d85a28f93312673a3 : SUCCESS in 42m 22s","accounts_in_message":[],"_revision_number":2},{"id":"9c816bc45945a127d5e7394f9097be20b81297ae","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-27 11:39:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a95b72be015c846649680d56418d45f372be0bb0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-28 14:48:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6d35a0f624776b9836ed8067e45f86c1588df458","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-28 14:48:27.000000000","message":"Topic changed from bug/1871840-stable/stein to bug/1871840","accounts_in_message":[],"_revision_number":2},{"id":"fb7846e55305ef9f99ef234ce842d8f92bbe3df0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 14:49:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"19404ee881d5709bbd244bf3368215772a9fe1cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 16:27:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e7fb680362d344df925356ffb29b9e96 : SUCCESS in 1h 25m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99490443baeb4d078dedab84b81fa29d : SUCCESS in 1h 23m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e351e738a0e41a1ab7713eb7594950e : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11c456c944ee4d0a9ad10ff60deb54af : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50e0e7ca76644188b82594223d93777e : SUCCESS in 9m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d3e77886189d4703a9cd1e86bc494caf : SUCCESS in 9m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e0487fb70eb423bba6dd9469189553a : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c9ad89d9dfa4e4aa30d2421a3faab0d : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/433028252ddf4d95b9675088f2f6a0fa : SUCCESS in 7m 36s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/361d9510ad0b46499e254e227ae49b1f : SUCCESS in 39m 35s","accounts_in_message":[],"_revision_number":2},{"id":"d0cadc04a181e34a68a3647c57cf8225e6e574a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 16:27:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2fb800d3f597295dbbfc103ddb3d0bb6c9ae8823","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 16:29:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c07523c848ca4b9bbd66b6ee61e261dd : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee96b0e346c0420e93938fe3d00cfeee : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"19fd9a2e2a283705552cb2c1f53c229b6ecf85e1","revisions":{"b97e81dad2eee2d334c3b5fe74c499e2493b9cda":{"kind":"REWORK","_number":1,"created":"2020-05-07 14:27:17.000000000","uploader":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"ref":"refs/changes/59/726159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/59/726159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/59/726159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/59/726159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/59/726159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/59/726159/1"}}},"commit":{"parents":[{"commit":"d5589a7366a60e396880dc2bf3fc0253f5feece6","subject":"Modify the constraints url in tox.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d5589a7366a60e396880dc2bf3fc0253f5feece6"}]}],"author":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-04-09 13:02:20.000000000","tz":120},"committer":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-05-07 14:27:17.000000000","tz":0},"subject":"Add missing SYSLOG_FACILITY to JournalHandler","message":"Add missing SYSLOG_FACILITY to JournalHandler\n\nWithout SYSLOG_FACILITY log messages from journald forwards the message\nto kern.log. This seems to be default behaviour as it treats the message\nas kernel message.\n\nThis option provides a proper facility to the log message and kern.log\nwill not be filled up anymore.\n\nChange-Id: I62a32ed46a400b62ead8c1e6e64acee658e2a769\nCloses-Bug: #1871840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b97e81dad2eee2d334c3b5fe74c499e2493b9cda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b97e81dad2eee2d334c3b5fe74c499e2493b9cda"}]},"branch":"refs/heads/stable/stein"},"19fd9a2e2a283705552cb2c1f53c229b6ecf85e1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-25 15:27:01.000000000","uploader":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"ref":"refs/changes/59/726159/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/59/726159/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/59/726159/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/59/726159/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/59/726159/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/59/726159/2"}}},"commit":{"parents":[{"commit":"d5589a7366a60e396880dc2bf3fc0253f5feece6","subject":"Modify the constraints url in tox.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d5589a7366a60e396880dc2bf3fc0253f5feece6"}]}],"author":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-04-09 13:02:20.000000000","tz":120},"committer":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-05-25 15:26:23.000000000","tz":120},"subject":"Add missing SYSLOG_FACILITY to JournalHandler","message":"Add missing SYSLOG_FACILITY to JournalHandler\n\nWithout SYSLOG_FACILITY log messages from journald forwards the message\nto kern.log. This seems to be default behaviour as it treats the message\nas kernel message.\n\nThis option provides a proper facility to the log message and kern.log\nwill not be filled up anymore.\n\nCherry-picked from commit 5b12e66f6283962d96884e4aa3fa7db7d4e062fa\nCherry-picked from commit 82751ce2a2b913a5d107133a15960c9c6d8fd594\nCherry-picked from commit 911a45401c1292b1946c200aa0f6a98b9e246fd8\nChange-Id: I62a32ed46a400b62ead8c1e6e64acee658e2a769\nCloses-Bug: #1871840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/19fd9a2e2a283705552cb2c1f53c229b6ecf85e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/19fd9a2e2a283705552cb2c1f53c229b6ecf85e1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
