)]}'
{"id":"openstack%2Foslo.log~726174","triplet_id":"openstack%2Foslo.log~stable%2Frocky~I62a32ed46a400b62ead8c1e6e64acee658e2a769","project":"openstack/oslo.log","branch":"stable/rocky","topic":"bug/1871840-stable/rocky","hashtags":[],"change_id":"I62a32ed46a400b62ead8c1e6e64acee658e2a769","subject":"Add missing SYSLOG_FACILITY to JournalHandler","status":"ABANDONED","created":"2020-05-07 15:19:36.000000000","updated":"2020-05-25 15:29:50.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d8507fbd2843600742955f4116178e029c708ec1","_number":726174,"virtual_id_number":726174,"owner":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-07 23:24:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-05-07 21:53:25.000000000","_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-1,"date":"2020-05-25 13:19:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 23:24:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-25 13:19:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"ea3433f70bea38ef4d515f2868e18fde77946b35","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-07 15:19:36.000000000","message":"Patch Set 1: Cherry Picked from branch stable/queens.","accounts_in_message":[],"_revision_number":1},{"id":"c5539bee69aa05f665c9ddaa51202cd25054e1b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 16:49:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e657a23e3d743c6b47e17d361390e66 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d26342a3841405fa535875362f2990a : SUCCESS in 6m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e7d0a20532f44909aef04b69d4e23274 : SUCCESS in 10m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d37020cb3e40aea7efe7d1025b19c9 : SUCCESS in 5m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f45c2ad6559142539260db11b61d0a47 : FAILURE in 1h 23m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83234c2e400b4c44a0840fe9eeca2675 : SUCCESS in 4m 48s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/07af5965e97642dc822164d4ff13d9a6 : SUCCESS in 1h 00m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7dda904507524d1ba6f7b177f639edad : SUCCESS in 8m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d2d21c74ec29c256684b0b7c1672a435140fc899","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-07 21:53:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"879dac64bacb547835409c5b7c20074607cae774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 23:24:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52935e71b0204081ad119cc5992d3949 : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6164bc9ee2346f696e1bdecf81fc287 : SUCCESS in 6m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f11667526f834d74ab629c5c1a63a485 : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d54773f61c424488a582322d8ae514 : SUCCESS in 5m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3f0ec3b91a8c4df2b35de4b832cb87fc : SUCCESS in 1h 25m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/822f559fe5464863bf50586ff9e0132c : SUCCESS in 7m 41s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/6def522334c049debffb0fdcc05c3ce8 : SUCCESS in 1h 12m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dc43a4460d1411991232db21b70eb08 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"688750a498dfe480a365f03eb5b66a7faf513b8d","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-09 18:08:35.000000000","message":"Topic set to bug/1871840-stable/rocky","accounts_in_message":[],"_revision_number":1},{"id":"779f0f95cf7a7d3b1df536bebeab6771916f9a3b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-25 13:19:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThanks\n\nSee my inline comment.\n\nAlso note that we stopped to release for `rocky` so we won\u0027t release this patch through a new version.\n\nhttps://github.com/openstack/releases/blob/master/deliverables/rocky/oslo.log.yaml#L36,L39","accounts_in_message":[],"_revision_number":1},{"id":"d8507fbd2843600742955f4116178e029c708ec1","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-05-25 15:29:50.000000000","message":"Abandoned\n\nno new release for rocky","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b17d7db9dc534584e06a75a136e39352577db28","revisions":{"3b17d7db9dc534584e06a75a136e39352577db28":{"kind":"REWORK","_number":1,"created":"2020-05-07 15:19:36.000000000","uploader":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"ref":"refs/changes/74/726174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/74/726174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/74/726174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/74/726174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/74/726174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/74/726174/1"}}},"commit":{"parents":[{"commit":"7900aa42a22ea8962fb33af680f636c20bf39119","subject":"Use upper constraints from opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/7900aa42a22ea8962fb33af680f636c20bf39119"}]}],"author":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-04-09 13:02:20.000000000","tz":120},"committer":{"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","date":"2020-05-07 15:19:36.000000000","tz":0},"subject":"Add missing SYSLOG_FACILITY to JournalHandler","message":"Add missing SYSLOG_FACILITY to JournalHandler\n\nWithout SYSLOG_FACILITY log messages from journald forwards the message\nto kern.log. This seems to be default behaviour as it treats the message\nas kernel message.\n\nThis option provides a proper facility to the log message and kern.log\nwill not be filled up anymore.\n\nChange-Id: I62a32ed46a400b62ead8c1e6e64acee658e2a769\nCloses-Bug: #1871840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3b17d7db9dc534584e06a75a136e39352577db28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3b17d7db9dc534584e06a75a136e39352577db28"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
