)]}'
{"id":"openstack%2Foslo.log~732457","triplet_id":"openstack%2Foslo.log~master~I5269b82d219fd2377535120d9d266238d50431b3","project":"openstack/oslo.log","branch":"master","topic":"bug/1871840","hashtags":[],"change_id":"I5269b82d219fd2377535120d9d266238d50431b3","subject":"Default facility to None in OSJournalHandler class","status":"MERGED","created":"2020-06-01 21:58:34.000000000","updated":"2020-06-03 09:55:05.000000000","submitted":"2020-06-02 19:32:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732457-1591126365814-6506d538","meta_rev_id":"60c4f61ba6edf1f14f02632f56db07e6973751ef","_number":732457,"virtual_id_number":732457,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-02 19:32:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2020-06-02 14:09:23.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-06-02 18:00:52.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-02 06:13:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"value":1,"date":"2020-06-02 15:21:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2020-06-02 07:51:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},{"value":1,"date":"2020-06-02 06:03:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-02 07:18:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2020-06-02 18:00:52.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 06:03:52.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-06-02 06:13:16.000000000","updated_by":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"reviewer":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"state":"REVIEWER"},{"updated":"2020-06-02 07:18:02.000000000","updated_by":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"reviewer":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"state":"REVIEWER"},{"updated":"2020-06-02 07:51:39.000000000","updated_by":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"reviewer":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"state":"REVIEWER"},{"updated":"2020-06-02 14:09:23.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-06-02 15:21:20.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-06-02 18:00:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-02 19:32:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b701910f0acdf6e2a330f4129bd4b40ded180b5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-01 21:58:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"019cd49346b31076030ca09868822ed65db4c69b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 23:30:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac349303593048729a7a8d008555c28d : SUCCESS in 1h 31m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3109a462a3f846499959df1c035e9568 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68b04b87cb294fbca5f778cad3207587 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ea7dfe11cb348abb938397cbb282aaf : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7945577ecf904d7d87e073ef49441fa7 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5985d474097d4c4cadf54f8b8933a0cc : SUCCESS in 3m 20s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/951f47fae2e245039303f7e1d2596aa7 : SUCCESS in 39m 41s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/5098dd1e0e844264afa567b453c26d5e : SUCCESS in 39m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2517f31a137998ee86aad3eaa578eb2365ef8c99","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-06-02 06:03:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bf8de9b01f9260c8442b22551b265410c10e08a1","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2020-06-02 06:13:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d66b020b63a5a8f4eaea57dc79cb68bf140289e1","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-06-02 07:18:02.000000000","message":"Patch Set 1: Code-Review+1\n\nwould also need to be backported to ussuri/train/stein","accounts_in_message":[],"_revision_number":1},{"id":"8919ef1cf78cbc35e1a94e159b274556b4f3860e","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2020-06-02 07:51:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5ae0f17ddd1cffc34d479a35705f80750f8e92d6","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-06-02 14:09:23.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"a70b8d3a2b4025b1f30ef9b82eadb7298e5202e8","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-02 15:21:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c1eeff9748b94ea2e44521eb86c08ef79766d656","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-02 18:00:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35cac25eec174e1335b0185d74128916fcfe0b76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 18:01:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60d2b6f8e2bf41f4a4b44d4664798d5bb86e6620","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 19:32:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e53b1aa17ed48d79982a6c9b0028d30 : SUCCESS in 1h 30m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8b9c0f27e7b40c2b8bba67a102e7881 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28a6a7accfa1413ab7b0f95f2f79cb3c : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b546ddb548841598139474962b77641 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c431fbfbbbe24df38be08ff4bd71c5c5 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/438ab2adc68c4d58a585d610fc368cf9 : SUCCESS in 3m 39s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/c50f4b73412f4431919b2cc2c7bfd451 : SUCCESS in 42m 47s","accounts_in_message":[],"_revision_number":1},{"id":"20b3ff063b6e68c8ff309b36294485cb12061be1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 19:32:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"381194840590b9926457cf2b39d127777e8df7e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 19:35:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df7815af11064a4fa713388bd279e5c9 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"35e84160960fd8cac8587a757b6e71fc81e1be38","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-03 09:55:05.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit c77f3c9aa82364c482be9b37e6d29634273bb90f","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"184235c26517598d6e60f4db72e095ba061a0d6e","revisions":{"184235c26517598d6e60f4db72e095ba061a0d6e":{"kind":"REWORK","_number":1,"created":"2020-06-01 21:58:34.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/57/732457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/57/732457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/57/732457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/57/732457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/57/732457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/57/732457/1"}}},"commit":{"parents":[{"commit":"d0be4dcb937c266e7946281e9b6473a7180d9941","subject":"Merge \"Add missing SYSLOG_FACILITY to JournalHandler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d0be4dcb937c266e7946281e9b6473a7180d9941"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-06-01 21:25:56.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-06-01 21:58:33.000000000","tz":-240},"subject":"Default facility to None in OSJournalHandler class","message":"Default facility to None in OSJournalHandler class\n\nUse a default value of None for facility in OSJournalHandler\nclass, as is done in the OSSysLogHandler class.\n\nNeutron started failing with a:\n\n E1120: No value for argument \u0027facility\u0027 in constructor call\n\nWith a recent release of oslo.log.\n\nChange-Id: I5269b82d219fd2377535120d9d266238d50431b3\nRelated-Bug: #1871840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/184235c26517598d6e60f4db72e095ba061a0d6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/184235c26517598d6e60f4db72e095ba061a0d6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
