)]}'
{"id":"openstack%2Foslo.log~733084","triplet_id":"openstack%2Foslo.log~stable%2Fussuri~I5269b82d219fd2377535120d9d266238d50431b3","project":"openstack/oslo.log","branch":"stable/ussuri","topic":"bug/1871840","hashtags":[],"change_id":"I5269b82d219fd2377535120d9d266238d50431b3","subject":"Default facility to None in OSJournalHandler class","status":"MERGED","created":"2020-06-03 09:55:05.000000000","updated":"2020-06-03 12:45:07.000000000","submitted":"2020-06-03 12:43:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733084-1591188234280-4106ee0f","meta_rev_id":"298ff1835ec636596504f699f7d3b13bafbd638a","_number":733084,"virtual_id_number":733084,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 12:43:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-03 11:33:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-03 10:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-03 10:48:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-03 11:33:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2020-06-03 09:55:05.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 09:55:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-03 10:28:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-03 10:48:26.000000000","updated_by":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"reviewer":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"state":"REVIEWER"},{"updated":"2020-06-03 11:33:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-03 12:43:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"382c9a8507bd4384adae7214ca4da726fea5aaea","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-03 09:55:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c41ed056ccec4f886328cfdfeab04f83103b29a8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-03 09:55:12.000000000","message":"Topic changed from bug/1871840-stable/ussuri to bug/1871840","accounts_in_message":[],"_revision_number":1},{"id":"ab59e6cc8805aa66c75d3e51a225def115adfa1b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-03 09:55:39.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 2045f4b38fd6daee384014f114a7c077aab9bb3a","accounts_in_message":[],"_revision_number":1},{"id":"c4fe81fb8581e04d04c94e99e8cfd92236d89c5f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-03 10:28:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff6a526846cf03112074fa6bb0c200e385e7c29c","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-06-03 10:48:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a90a54ee94cb114a039664bc89b0230931151701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 11:10:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aab7ad2c41ad44a3a6c7c22c064cd36c : SUCCESS in 1h 12m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e41e0cde822a4d4dbdaab40c57c190c2 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/735529f949324324894b1588d7b313ab : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16f1eb2412064ae38e1928178868e0c5 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/570bf53727f54d409a0680ea235a884f : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eefd369730f45ac8bf1c9b2196142da : SUCCESS in 4m 21s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/d02576420a1c4470a913b4bd008d31d5 : SUCCESS in 1h 07m 20s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/5d66a2b5bb4e48f7aa712e34e5f91267 : SUCCESS in 39m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ad65a0a14f2f76bc58360ff022a18d0d9621501e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-03 11:33:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"07f29646398387669d590adceabcd76a37345f63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 11:33:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9fa7e5fa4a8af4818b74ff7b0782bb386f92d8ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 12:43:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a99eae85b164815aae95bb8aae3bd2c4d9d1eb05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 12:43:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb6f0061e77f42f4b0d0de5082885aef : SUCCESS in 1h 09m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5380a6312dd5446c8a6c0a365614f2e9 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66a570cc810144afaec4ce227421f03e : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d15512f0a1bd4497ac7cacc8f1bd70ed : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c04e6ede0664547a99819ed89804a29 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca1cd2972f894e6e93ea67e633dff5ac : SUCCESS in 3m 57s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/9e49a3e7704844c9b131b20d7db44f4f : SUCCESS in 43m 27s","accounts_in_message":[],"_revision_number":1},{"id":"298ff1835ec636596504f699f7d3b13bafbd638a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 12:45:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/909bd27297a14542b42b5ae727d51078 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c77f3c9aa82364c482be9b37e6d29634273bb90f","revisions":{"c77f3c9aa82364c482be9b37e6d29634273bb90f":{"kind":"REWORK","_number":1,"created":"2020-06-03 09:55:05.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/733084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/84/733084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/84/733084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/84/733084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/84/733084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/84/733084/1"}}},"commit":{"parents":[{"commit":"a67f7b5ce5b956918640784d9a180873676d5820","subject":"Add missing SYSLOG_FACILITY to JournalHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a67f7b5ce5b956918640784d9a180873676d5820"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-06-01 21:25:56.000000000","tz":-240},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-06-03 09:55:05.000000000","tz":0},"subject":"Default facility to None in OSJournalHandler class","message":"Default facility to None in OSJournalHandler class\n\nUse a default value of None for facility in OSJournalHandler\nclass, as is done in the OSSysLogHandler class.\n\nNeutron started failing with a:\n\n E1120: No value for argument \u0027facility\u0027 in constructor call\n\nWith a recent release of oslo.log.\n\nChange-Id: I5269b82d219fd2377535120d9d266238d50431b3\nRelated-Bug: #1871840\n(cherry picked from commit 184235c26517598d6e60f4db72e095ba061a0d6e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/c77f3c9aa82364c482be9b37e6d29634273bb90f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/c77f3c9aa82364c482be9b37e6d29634273bb90f"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
