)]}'
{"id":"openstack%2Foslo.log~733086","triplet_id":"openstack%2Foslo.log~stable%2Fstein~I5269b82d219fd2377535120d9d266238d50431b3","project":"openstack/oslo.log","branch":"stable/stein","topic":"bug/1871840","hashtags":[],"change_id":"I5269b82d219fd2377535120d9d266238d50431b3","subject":"Default facility to None in OSJournalHandler class","status":"MERGED","created":"2020-06-03 09:57:17.000000000","updated":"2020-06-05 17:51:41.000000000","submitted":"2020-06-05 17:50:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733086-1591379416647-d5a688fa","meta_rev_id":"52e2cdf2441f5fb78842e5cfdbce4639a86c5d54","_number":733086,"virtual_id_number":733086,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"date":"2020-06-03 12:46:33.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 17:50:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-03 16:47:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-06-03 15:29:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-03 10:48:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":0,"date":"2020-06-03 12:47:44.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-05 15:56:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2020-06-03 09:57:17.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 09:57:17.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-03 10:48:35.000000000","updated_by":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"reviewer":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"state":"REVIEWER"},{"updated":"2020-06-03 12:47:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-03 15:29:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-05 15:56:24.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-06-05 17:50:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0466d723769d939657bb50d2c911707a066617f4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-03 09:57:17.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"f22fa89483760969a24ab63a3148cdf08db1a34e","author":{"_account_id":31823,"name":"Benjamin Reichel","email":"b.reichel@syseleven.de","username":"benj200"},"date":"2020-06-03 10:48:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c842989027057c86dd5b84be82615c22ed704da5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 11:52:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/75ce6ef674424e65a4b8954577d9ffa9 : SUCCESS in 1h 33m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92587f9bc4764f99acff4bfba2958ed1 : FAILURE in 1h 54m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0b63d27408c467fb42a589ebb370b8f : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de85893e47b84bd5bba20b25029d4e58 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/882bc5e4fc5b4a9b86653bafb105be7b : SUCCESS in 3m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/56df71a3c29443ebbc7dcc83a82f347c : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/771dbaadfa624541ba7ba4f238958891 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a479f1cb8da4c47922c4e33976e64d7 : SUCCESS in 5m 06s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/11d33948288a450ca9de378d8d281d61 : SUCCESS in 53m 19s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/b00570346a9549328cb2dfea5c5784df : SUCCESS in 27m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bc737d06879cb71198cebac7ebcb83df674e1c92","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-03 12:46:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc010dc3cf8e38019ed330d5ec3f7cd01d0d5c0b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-03 12:47:55.000000000","message":"Topic changed from bug/1871840-stable/stein to bug/1871840","accounts_in_message":[],"_revision_number":1},{"id":"2b2a8135635d8d3e8951e234f9e7b8953230afc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 14:16:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ce1d765b59d94bfd8949ed4e14fcd2e0 : SUCCESS in 1h 25m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c501913277264307a52f2f4614b57c36 : SUCCESS in 1h 25m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b7cda349abc4eb190126a0bec00636a : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f203de8c350843d7ae69931aae6f5d7b : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0834b3545e0f4f9088ee230ad8444584 : SUCCESS in 4m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cc3e04d3882349aeb8775746ec7866ca : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d694af6990ce40238bea93b60d4aca9c : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cd9db699b9b440282c67c51b04b0039 : SUCCESS in 4m 37s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/ac873d58038e4a7d87420f5a2b63189e : SUCCESS in 1h 00m 19s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/33f19463f1ae4c2cb34834a781b0f0a3 : SUCCESS in 38m 32s","accounts_in_message":[],"_revision_number":1},{"id":"816482e449e07c569631abf6b037caa6672911cb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-03 15:29:56.000000000","message":"Patch Set 1: Code-Review+2\n\nClean backport","accounts_in_message":[],"_revision_number":1},{"id":"7ee5a2a89f97ec30aa8b37c5065512c2130fd4dd","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-06-03 16:47:52.000000000","message":"Patch Set 1: Code-Review+2\n\nWaiting on train backport to merge.","accounts_in_message":[],"_revision_number":1},{"id":"d4be8be037d66d547120b29e744fdb89f87e2c75","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-06-05 15:56:24.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a4e4f38d3dc10e47b3588fb9a2c1374d9f3bab2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:56:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3dbd1589f37a17de2d590ec8fdef9c06b9b07004","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 17:50:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"903da0bc1ac896bc9067301581f7210ae4fd3950","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 17:50:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e15dfd7962ca46908a26a02f0e4e413d : SUCCESS in 1h 02m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be2ec34d9107491c8781ff7a3bd8a2f4 : SUCCESS in 1h 25m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/faad920a3e5d4d229ccae91341ff4a3f : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3218f4907134d6395559bd5a0de212d : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70496c7569a94ffdb6bbe62925b1176e : SUCCESS in 4m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3514d895f0654585ac08046a1c8f4d4f : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd9c2291b1ad40db82224947a2d73959 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d38e4f4cbc4502b7109f1acc2e3134 : SUCCESS in 4m 52s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/6b5160e952e242f3a70cdaef8ab3d84e : SUCCESS in 44m 56s","accounts_in_message":[],"_revision_number":1},{"id":"52e2cdf2441f5fb78842e5cfdbce4639a86c5d54","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 17:51:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6676ff2f8ee4a1b835e5c7d14c56a2a : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef23f7085059a7844be308db454a20168fbc97e1","revisions":{"ef23f7085059a7844be308db454a20168fbc97e1":{"kind":"REWORK","_number":1,"created":"2020-06-03 09:57:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/733086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/86/733086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/733086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/733086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/733086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/86/733086/1"}}},"commit":{"parents":[{"commit":"19fd9a2e2a283705552cb2c1f53c229b6ecf85e1","subject":"Add missing SYSLOG_FACILITY to JournalHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/19fd9a2e2a283705552cb2c1f53c229b6ecf85e1"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-06-01 21:25:56.000000000","tz":-240},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-06-03 09:57:17.000000000","tz":0},"subject":"Default facility to None in OSJournalHandler class","message":"Default facility to None in OSJournalHandler class\n\nUse a default value of None for facility in OSJournalHandler\nclass, as is done in the OSSysLogHandler class.\n\nNeutron started failing with a:\n\n E1120: No value for argument \u0027facility\u0027 in constructor call\n\nWith a recent release of oslo.log.\n\nChange-Id: I5269b82d219fd2377535120d9d266238d50431b3\nRelated-Bug: #1871840\n(cherry picked from commit 184235c26517598d6e60f4db72e095ba061a0d6e)\n(cherry picked from commit c77f3c9aa82364c482be9b37e6d29634273bb90f)\n(cherry picked from commit e19c4076b1f0d7fdbd6d68a09c973934029926a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ef23f7085059a7844be308db454a20168fbc97e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ef23f7085059a7844be308db454a20168fbc97e1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
