)]}'
{"id":"openstack%2Foslo.log~740552","triplet_id":"openstack%2Foslo.log~master~I158409677523978246c26bf1168f323974d817a5","project":"openstack/oslo.log","branch":"master","topic":"bug/1887232","hashtags":[],"change_id":"I158409677523978246c26bf1168f323974d817a5","subject":"Added uwsgi_name information into fluentFormatter event message","status":"MERGED","created":"2020-07-11 14:58:42.000000000","updated":"2020-08-06 16:24:28.000000000","submitted":"2020-08-06 16:20:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740552-1596730805950-afcfe1c4","meta_rev_id":"bc318f52cdd4c011b78f2b9ccc538846ad756d81","_number":740552,"virtual_id_number":740552,"owner":{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-15 10:46:31.000000000","_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 16:20:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},{"value":2,"date":"2020-08-06 15:01:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-07-15 11:39:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},{"value":1,"date":"2020-08-06 15:01:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-15 11:39:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-06 15:01:22.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-08-06 16:20:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d36b4158a81d9301f9aca6a9c536c3163ac1977f","author":{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},"date":"2020-07-11 14:58:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d0098dbd71f1e89c0e61b87671686168a723256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 15:58:10.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fededa7c5b854aa4a16d829b6501cda1 : SUCCESS in 57m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d148cf60b5c48eb94d25b8ce81d3e97 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/359c6ba92f1d4654a4575948d3ab95a5 : FAILURE in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b7cbec1535e4d6f893aecd006b644f1 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/071e77e95c1248d08380cbe68b80d076 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81a3bacaf37944cda9da0d9aa8dfe01d : SUCCESS in 3m 31s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/c44b2c54142943748f6af4a6840d9d5f : SUCCESS in 58m 41s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/c5a7df0bac3a4d2a9e8e1bce1f7aa910 : SUCCESS in 39m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8d61456b05e029ee0c1277ad0fff957bd9ba9035","author":{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},"date":"2020-07-11 16:01:41.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"60ecc3bb38d0abe544d46a3e25f6792c2246a937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 17:30:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b58daa676d154db48b6afbb22d0efef9 : SUCCESS in 1h 28m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d26625439d94b10807964ba110ad4ae : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/117b14ca0c424700bf262b16993af0e1 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a921a715ad54f2f8d328844ec24d6c9 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0955fa5dc6ea4200a9fbd06d19e3726c : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ef22fb505e4dbfb0bd3d8c992a5cdb : SUCCESS in 3m 11s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/d9ef4850e0914489acc62c04ea4634db : SUCCESS in 55m 55s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/41e5d51a3003403c9ab248044893d42f : SUCCESS in 35m 37s","accounts_in_message":[],"_revision_number":2},{"id":"2d588b452492fcaa45c862d1aa598413f537769e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-15 10:04:11.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"80a6c2c319170c261e8789d9a2e0221aeecef11f","author":{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},"date":"2020-07-15 10:15:24.000000000","message":"Patch Set 2:\n\n(2 comments)\n\n\u003e (2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2f4e47fac84977cc6579e6a52d53616fc5b302a4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-15 10:26:55.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fc54be481a66e6cc839c2b3cda88c4b31cfcc2b3","author":{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},"date":"2020-07-15 10:46:31.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e (2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"23dfb0fbd789608a0e74d804a1251c17d506f221","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-15 11:39:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"25b4c74fee4df36b85618682d7bc648d6acd7fb7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-06 15:01:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"70b3454fd8097b3d2c0b1fb804cad85d9fe27990","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 15:01:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b31f829ef37b73d2c9849a39b814941271a4e835","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:20:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fe0b676a474077e7c63f5585ed0ac6fe0ebcefaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:20:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74ff58a339274bfabafd628770ac2377 : SUCCESS in 58m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/669958cb35f14b73a93ac4008f3483c3 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57352ba8ce554d64804ff6e4a4233ff1 : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1def2824802544058a491885a8f34cd9 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27d9d523c72e46f0b5ffeb5dc52f6fab : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0acdb63e0c3497d9ecf8691ed9fe4dd : SUCCESS in 4m 40s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/a3cb01281643478b9d1ff3fc378a81d3 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":2},{"id":"9886e06c3d4e0927e90c8f7c97cab5222f42fd3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:24:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740e37cfba42403e9191f14a5b3f52c7 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f9b434bb3bc4014d5041f6b1a1da858e3d4aa77","revisions":{"7e21ebe4a810adf14a9f5ce63ce5a7fdf611bab3":{"kind":"REWORK","_number":1,"created":"2020-07-11 14:58:42.000000000","uploader":{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},"ref":"refs/changes/52/740552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/52/740552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/52/740552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/52/740552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/52/740552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/52/740552/1"}}},"commit":{"parents":[{"commit":"0a2dc7096bcdb83431e0c6e9b73800540596956e","subject":"Merge \"Align contributing doc with oslo\u0027s policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0a2dc7096bcdb83431e0c6e9b73800540596956e"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2020-07-11 14:58:29.000000000","tz":120},"committer":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2020-07-11 14:58:29.000000000","tz":120},"subject":"Added uwsgi_name information into fluentFormatter event message","message":"Added uwsgi_name information into fluentFormatter event message\n\nWhen running under uwsgi a new key \"uwsgi_name\" is added to log\nevents when using FluentFormatter that could be used for further\nanalysis. It\u0027s value is taken from uwsgi configuration upon\nformatter initialization.\n\nChange-Id: I158409677523978246c26bf1168f323974d817a5\nCloses-Bug: 1887232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/7e21ebe4a810adf14a9f5ce63ce5a7fdf611bab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/7e21ebe4a810adf14a9f5ce63ce5a7fdf611bab3"}]},"branch":"refs/heads/master"},"6f9b434bb3bc4014d5041f6b1a1da858e3d4aa77":{"kind":"REWORK","_number":2,"created":"2020-07-11 16:01:41.000000000","uploader":{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},"ref":"refs/changes/52/740552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/52/740552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/52/740552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/52/740552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/52/740552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/52/740552/2"}}},"commit":{"parents":[{"commit":"0a2dc7096bcdb83431e0c6e9b73800540596956e","subject":"Merge \"Align contributing doc with oslo\u0027s policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0a2dc7096bcdb83431e0c6e9b73800540596956e"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2020-07-11 14:58:29.000000000","tz":120},"committer":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2020-07-11 16:01:32.000000000","tz":0},"subject":"Added uwsgi_name information into fluentFormatter event message","message":"Added uwsgi_name information into fluentFormatter event message\n\nWhen running under uwsgi a new key \"uwsgi_name\" is added to log\nevents when using FluentFormatter that could be used for further\nanalysis. It\u0027s value is taken from uwsgi configuration upon\nformatter initialization.\n\nChange-Id: I158409677523978246c26bf1168f323974d817a5\nCloses-Bug: 1887232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6f9b434bb3bc4014d5041f6b1a1da858e3d4aa77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6f9b434bb3bc4014d5041f6b1a1da858e3d4aa77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
