)]}'
{"id":"openstack%2Foslo.log~742135","triplet_id":"openstack%2Foslo.log~master~I6663d5e0866e03f8eb8c0543d3e3e0bbfb2399ac","project":"openstack/oslo.log","branch":"master","topic":"oslo-pre-commit","hashtags":[],"change_id":"I6663d5e0866e03f8eb8c0543d3e3e0bbfb2399ac","subject":"Adding pre-commit","status":"MERGED","created":"2020-07-21 11:32:43.000000000","updated":"2020-09-22 12:37:42.000000000","submitted":"2020-09-22 12:34:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742135-1600778073466-57e98530","meta_rev_id":"79dcf57d1c5c110598c8bd8912f97ded43992299","_number":742135,"virtual_id_number":742135,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-22 12:34:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":2,"date":"2020-09-08 14:46:00.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":2,"date":"2020-09-22 11:15:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":1,"date":"2020-09-22 11:15:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 14:46:00.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2020-09-22 11:15:06.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-09-22 12:34:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea2a09ad7168ffdcbf17cb0d59fa138803a21c7d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-21 11:32:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f64f7de077a3e146eb3b5475e1f8e03b9707f9e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 13:14:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69beac3bbe144cafb662ad4249605a97 : SUCCESS in 1h 33m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b06f46e462664a4fa63aa00355129fa8 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ada6406e18584440a4d93ea6d2bc30a4 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5562ea6ddbe54fa295936a3e74791dba : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1bc40be73c04a138946efd40039b125 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2838b816a2b481cb1de142c427f7353 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10e41494c42c403fba21dc2fc2a282fd : SUCCESS in 5m 05s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/3c5be166311a4a86a0a1ebb17aae45c1 : SUCCESS in 54m 14s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/f344fffed4df4b7b98feab32214b8f6a : SUCCESS in 38m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c507a87639e4db62487322ec744bfa9e0ee8e997","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-07-22 08:40:42.000000000","message":"Patch Set 1: Code-Review-2\n\nWe should hold on in these patches for a while and observe oslo.cache for some time before spreading the change around all Oslo.","accounts_in_message":[],"_revision_number":1},{"id":"c8131d87ef9d4a65a6fe8a66185280a71fe20a2a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-08 12:21:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"910b5d941fe8371ac0cd383d9c57801d41cb4a6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 13:54:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a71aadca2d6d48dc888cf75a4c05dfbf : SUCCESS in 2m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba6043a75a3d49e6aeabcbd6939fdfc6 : SUCCESS in 1h 29m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bfa802a488a47b0a430596db7141c28 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ac1b0816374abd98a2d70f63a2cdb9 : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5adfb881d394cf794e5fc0189a136d2 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79142447077b4de188805f63b91a6c1c : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e71cebdb1a640078f991cac5e47879b : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a1861a27338431bbf25f7d0564bc7fb : SUCCESS in 3m 12s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/5e40dcaaafdb417e9aee26866d02f8c8 : SUCCESS in 1h 00m 16s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/e190eef3c62049148d125ba132a36409 : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":2},{"id":"1cd6a4782bb0142bea024dd431483aa0f7122d3b","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-08 14:45:45.000000000","message":"Removed Code-Review-2 by Moisés Guimarães \u003cmoguimar@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"6d0489afc1ef982fa530be67d07d0cf853a9f52e","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-08 14:46:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2934276dec2b696848ccadb62d90c1d0622c4ae0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-22 11:15:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5e6e781971d3026acb9c60237dfe2908b4830252","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 11:15:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"516d9f0e4a9e957557cfe338a63a981b9ec70e22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 12:34:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f876129292021c6cd7d3cca7de9206a2c2026724","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 12:34:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6d6ba28ce32e4cb7a907e572d806edcc : SUCCESS in 3m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94cae6f12cfa46f8a7d9d88cc6c2570e : SUCCESS in 1h 12m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69605f80dc22483caaa56c820685e213 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c5cd9f1803f409eac6439c3602b8359 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe7f192659344f25bd7b0ef48b55362c : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b37bc724107c4fc6891b8f93c79c9b11 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea43caf6109946838856e3824b2a1673 : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aec6f0c74bff4de0a6348b55d28a5bae : SUCCESS in 4m 01s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/7b2e6cf9e6784fd3bac1105e57ce5b80 : SUCCESS in 42m 52s","accounts_in_message":[],"_revision_number":2},{"id":"1304500545d38f7cb6a2f4a374de1d360182af46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 12:37:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02dbbaa4ea8143a5b1ee343f87bedc3d : SUCCESS in 2m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e725ba56513425a99ed6e0acec1a0fc : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ad42c89f3be22c6e061c2da52bcd8cecdb665bc8","revisions":{"7cac01e3024f6b2040d0f336741aaedcb12606b0":{"kind":"REWORK","_number":1,"created":"2020-07-21 11:32:43.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/35/742135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/35/742135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/35/742135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/35/742135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/35/742135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/35/742135/1"}}},"commit":{"parents":[{"commit":"0a2dc7096bcdb83431e0c6e9b73800540596956e","subject":"Merge \"Align contributing doc with oslo\u0027s policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0a2dc7096bcdb83431e0c6e9b73800540596956e"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:32:24.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:32:24.000000000","tz":120},"subject":"Add pre-commit","message":"Add pre-commit\n\npre-commit will be used to:\n- trailing whitespace [1].\n- Replaces or checks mixed line ending (mixed-line-ending) [1].\n- Forbid files which have a UTF-8 byte-order\n  marker (check-byte-order-marker) [1].\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs) [1].\n- Check for files that contain merge conflict\n  strings (check-merge-conflict) [1].\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements) [1].\n- Attempts to load all yaml files to verify syntax (check-yaml) [1].\n- replace tabs by space (4 by default)(remove-tabs) [2].\n- Run flake8 checks (flake8) (local)\n\n[2] https://github.com/pre-commit/pre-commit-hooks\n[3] https://github.com/Lucas-C/pre-commit-hooks\n\nChange-Id: I6663d5e0866e03f8eb8c0543d3e3e0bbfb2399ac\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/7cac01e3024f6b2040d0f336741aaedcb12606b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/7cac01e3024f6b2040d0f336741aaedcb12606b0"}]},"branch":"refs/heads/master"},"ad42c89f3be22c6e061c2da52bcd8cecdb665bc8":{"kind":"REWORK","_number":2,"created":"2020-09-08 12:21:58.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/35/742135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/35/742135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/35/742135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/35/742135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/35/742135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/35/742135/2"}}},"commit":{"parents":[{"commit":"9b29c909dbb418e75b5841de09d0df0bae7c91d1","subject":"Merge \"Added uwsgi_name information into fluentFormatter event message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/9b29c909dbb418e75b5841de09d0df0bae7c91d1"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:32:24.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-08 12:21:18.000000000","tz":120},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules.\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker\n  (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings\n  (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I6663d5e0866e03f8eb8c0543d3e3e0bbfb2399ac\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ad42c89f3be22c6e061c2da52bcd8cecdb665bc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ad42c89f3be22c6e061c2da52bcd8cecdb665bc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
