)]}'
{"id":"openstack%2Foslo.log~744313","triplet_id":"openstack%2Foslo.log~master~Ib0224e9eca76483d7b60007cb96c71703c8a5332","project":"openstack/oslo.log","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"Ib0224e9eca76483d7b60007cb96c71703c8a5332","subject":"DNM: testing gate on ubuntu focal","status":"ABANDONED","created":"2020-08-01 22:34:47.000000000","updated":"2020-08-02 23:27:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bcbc56effb71945e1e6507b09d92bd1bbd7978ea","_number":744313,"virtual_id_number":744313,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-02 21:25:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-02 21:25:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac3dc5f4b069326a852250e6fc00442e4cf0ee76","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-01 22:34:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f361038ae1f0ed785fc00bbe31a3a86460d3a23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 23:35:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cff051e667214482b0c4da5279e365b2 : SUCCESS in 58m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6515667482d427d9c7efddc4fdeea01 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eed3584aa31c4782a1d233e4dc3c4035 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ced9f0ca367d49e9b50ae3b084f1c543 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56d3a1c7695b418e86d0cea0ddc8226c : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abaddb28ab6042c8b781c3ba0b44cec7 : SUCCESS in 2m 59s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/54e1cd7b33684aea9c1ac1ab969e6dca : SUCCESS in 56m 11s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/7627954e396943cc83de906bee1f6d24 : SUCCESS in 40m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7515fc0049ed4b8e9995adb90e58e64ac42e52a2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 19:47:59.000000000","message":"Abandoned\n\nall working here, no updates needed for Focal migration.","accounts_in_message":[],"_revision_number":1},{"id":"b771153ea71ebba8247bc5829296a63b02eac636","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 19:49:40.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"c3aa5d2f625c8410392452c04b39d27a478a3764","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 19:50:12.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"036362659876dad9862d6f68213f4c2380ea8bfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 21:25:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9f25053af4a43bc906eb48c99b0583e : SUCCESS in 1h 27m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ac1c7323b1c4e5baf9892a43a1e1920 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/599df463f9a54c2eb804af943d233353 : SUCCESS in 2m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1b4771fd66c47be9f30abce9d340758 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea4f3ceb1e2141ba8afba8f82586d8c5 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a47f56bbd5684af891d2a7cc47bd8382 : SUCCESS in 3m 38s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/fa75f036c03d47209d73a0c2a6a7316a : SUCCESS in 54m 38s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/92ddf2eea57045e488ddcd0cce1c917d : SUCCESS in 26m 50s","accounts_in_message":[],"_revision_number":2},{"id":"d52d5a4eed17fd7034ee47cbfa77e4adc48b8f18","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 23:27:55.000000000","message":"Abandoned\n\nall working here, no updates needed for Focal migration.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e474498863ceea5a1f2c233bccddd30da1edeeeb","revisions":{"778eeaf30dd8ade6bd44886f3aa777c6a7f09054":{"kind":"REWORK","_number":1,"created":"2020-08-01 22:34:47.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/13/744313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/13/744313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/13/744313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/13/744313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/13/744313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/13/744313/1"}}},"commit":{"parents":[{"commit":"0a2dc7096bcdb83431e0c6e9b73800540596956e","subject":"Merge \"Align contributing doc with oslo\u0027s policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0a2dc7096bcdb83431e0c6e9b73800540596956e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 22:31:58.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 22:34:46.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nChange-Id: Ib0224e9eca76483d7b60007cb96c71703c8a5332\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/778eeaf30dd8ade6bd44886f3aa777c6a7f09054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/778eeaf30dd8ade6bd44886f3aa777c6a7f09054"}]},"branch":"refs/heads/master"},"e474498863ceea5a1f2c233bccddd30da1edeeeb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-02 19:50:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/13/744313/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/13/744313/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/13/744313/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/13/744313/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/13/744313/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/13/744313/2"}}},"commit":{"parents":[{"commit":"0a2dc7096bcdb83431e0c6e9b73800540596956e","subject":"Merge \"Align contributing doc with oslo\u0027s policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0a2dc7096bcdb83431e0c6e9b73800540596956e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 22:31:58.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 19:50:10.000000000","tz":0},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nChange-Id: Ib0224e9eca76483d7b60007cb96c71703c8a5332\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e474498863ceea5a1f2c233bccddd30da1edeeeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e474498863ceea5a1f2c233bccddd30da1edeeeb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
