)]}'
{"id":"openstack%2Foslo.log~749191","triplet_id":"openstack%2Foslo.log~stable%2Fqueens~Idc0b9edc61b4ec39dc9df0ea8282cc104e4d83c0","project":"openstack/oslo.log","branch":"stable/queens","topic":"queens/bug/1802142","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-11-30 16:24:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idc0b9edc61b4ec39dc9df0ea8282cc104e4d83c0","subject":"Fix handling of exc_info in OSJournalHandler","status":"ABANDONED","created":"2020-09-01 07:18:58.000000000","updated":"2022-11-30 16:24:32.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"068ed77788fcf5436e6e685bda81bbbd0205b863","_number":749191,"virtual_id_number":749191,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":9701,"name":"Nikita Gerasimov","email":"nikita.gerasimov@oracle.com","username":"nikitych"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-08 09:10:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-02-12 13:41:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9701,"name":"Nikita Gerasimov","email":"nikita.gerasimov@oracle.com","username":"nikitych"},{"value":2,"date":"2020-09-14 16:06:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":1,"date":"2021-02-12 13:41:18.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":9701,"name":"Nikita Gerasimov","email":"nikita.gerasimov@oracle.com","username":"nikitych"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"date":"2020-09-01 07:18:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9701,"name":"Nikita Gerasimov","email":"nikita.gerasimov@oracle.com","username":"nikitych"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"REVIEWER":[{"_account_id":9701,"name":"Nikita Gerasimov","email":"nikita.gerasimov@oracle.com","username":"nikitych"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 07:18:58.000000000","updated_by":{"_account_id":9701,"name":"Nikita Gerasimov","email":"nikita.gerasimov@oracle.com","username":"nikitych"},"reviewer":{"_account_id":9701,"name":"Nikita Gerasimov","email":"nikita.gerasimov@oracle.com","username":"nikitych"},"state":"REVIEWER"},{"updated":"2020-09-01 08:53:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-14 16:06:33.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-02-12 13:41:18.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2021-03-23 20:58:59.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2022-02-10 13:13:53.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"}],"messages":[{"id":"229a65536b3b3a0465f788ec538a26d921e40779","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-01 07:18:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fff16ba6b229c1a0c50052243f1ab25179dbb8b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 08:53:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/666eff70921a4ba8abeff021c2fd9eee : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17ad48c2171747c4aecada1c6b640749 : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/36daaca3709149c3a20b2d98f4149e72 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3976160b31a341c89c25e1685951fe73 : SUCCESS in 5m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/decc6d0a192e4b69b6a15075612ea00e : SUCCESS in 1h 21m 41s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/853cb9d3afec44649e4527e7b553ff18 : FAILURE in 8m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e62a2ffb2943232e352417cf05e2957398b6e50","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-14 16:06:33.000000000","message":"Patch Set 1: Code-Review+2\n\nstable/rocky is approved and on its way through the gate. Makes sense to me","accounts_in_message":[],"_revision_number":1},{"id":"c030f5c9944f7fb07d355a99184208521b667dc8","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-02-12 13:41:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36838a0813fae76c8183c90a3843cb601099322a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 13:41:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b91119567ca936b3c04b8bf83a5eab3169b1023f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 15:14:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad8873d25fa4e06bfb44ddbf301414b : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ea8dc5f98a649789fb1db8d4b3f3858 : FAILURE in 3m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/00d204adb564414cae9d67046cb84d0d : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b933de7f7604b6086b1bac8424ceec3 : SUCCESS in 4m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fce6c00d88e0425786d445cff0adde22 : SUCCESS in 1h 27m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a9f1f33c16c833a4061efdf4bd0f272b89530631","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 08:54:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"335f4715fc32d009f00df670e8381816fe9a5328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 09:10:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ccdad68f0b4c1eb859f3617b340b4a : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32fd9fa481a9487483c8b26ebc01dc09 : FAILURE in 3m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/66f83310c62340a2b399c54d7aef2cde : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c0b38801480450b9b3cf4f01b411bdc : SUCCESS in 5m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b2fe84366a974b38a287827465922fb1 : FAILURE in 11m 34s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/513c4e568b2e4b4b9acba77111090fe3 : FAILURE in 13m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"75e5247666aef5c2cfa709a93841659a736cd44e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 10:23:34.000000000","message":"Patch Set 1:\n\nThe py27 error will be fixed by \u003d\u003e https://review.opendev.org/c/openstack/oslo.log/+/779182","accounts_in_message":[],"_revision_number":1},{"id":"4243751d374b21b94a5f75d3f4b748e099de1268","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-23 20:58:59.000000000","message":"Patch Set 1:\n\nrecheck - gate is fixed","accounts_in_message":[],"_revision_number":1},{"id":"85742cffbe40d93743aa41bec96955a9ef6931c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 21:15:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9fb93c4524c4b1a8e3923d28a749424 : FAILURE in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02297e79cd864de28e3f71a5fd648956 : FAILURE in 2m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/22818fda1d704938a578dea027195ff5 : FAILURE in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/def95a0b346444e781d1af6c12801c56 : FAILURE in 4m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ef0295fae2064c989071a4239f135002 : FAILURE in 14m 36s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/cebbeaffc9b34738b746166148a798b2 : FAILURE in 8m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7386d7a95b952bf705e6e211d951704c89059d62","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-27 08:24:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f66b7e8ddbf5e2e33b54dc8868cfd81f2307893c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 08:44:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43f23397ddeb44029039693112e57759 : FAILURE in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa1896513db149639c8eb1e6def025c4 : FAILURE in 4m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/87fe9ba8e5134710b66f527b12c1dd89 : FAILURE in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8352bfafe18d4ee1aaeede18170d5d4a : FAILURE in 3m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3342485d0b164616b761c16c3eac3380 : FAILURE in 19m 11s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/dd369ed4e12147a49002bfa887f54f24 : FAILURE in 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3193dadda7de151b111b98d1525a131e72a4cb1b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-16 09:58:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9ff872be85815e04978dcd8f1afba64ad51a7519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 10:21:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5a63e5963cf497aacb93336cee89282 : FAILURE in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f831edd94734e0d81d718a011d22db7 : FAILURE in 5m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/85eb5cdea194428e83f3c205d6cffc6d : FAILURE in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6bf2fd009a048239807238f23d0c5a8 : FAILURE in 4m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ee324fbc0521483695e3544a008f8c73 : FAILURE in 19m 10s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/09ea68eb0e564a8fa6a20bb2187a127d : FAILURE in 20m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"393ef1e9a38c45fce2cd7aaddad634c0aeeadd93","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-02-10 13:13:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1f8b50d652489a8b642dfebbc6429e5d45c8f66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 13:48:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ed7e294f62e436eb02b754c66bf8352 : FAILURE in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f08b45cf68948e6bb1816fbc63237ce : FAILURE in 3m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/095b0d39170a4c62b0050e48793f116c : FAILURE in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfea254ff58a4b20ad43474f7dc10002 : FAILURE in 3m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/63830a2db8df478c8ecdcd18821af3f5 : FAILURE in 23m 03s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/9d5ee6c276e84fdd8322983f7a4d5298 : FAILURE in 26m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cede86a6d5f7a5bf3ed730f7066d43119dae58e9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-11 13:43:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3823d020611061500748cf0269258edab81ed17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 14:23:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e59a139e5fd4ce0a85c7fbbaa55d90b : FAILURE in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f87123efba17407cbe9e5d2ac1ea30f1 : FAILURE in 3m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6e19764e9a36485fb1b7cdc6f93940c0 : FAILURE in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e6b69e21703405fbf546d2cbfdbd1e6 : FAILURE in 3m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/47150e62c91a4adda4e12c09c09cc2b9 : FAILURE in 14m 26s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/705404f94c7549b3aca23089164eaea8 : FAILURE in 32m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a73bc2792be417f05eef460fdea9f934c4a05cb6","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-04-25 12:43:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6245c6c1c8c61be56c1aae5c127c823b6385dee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 13:12:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b51911758754891ab6923d9706e9256 : FAILURE in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0ddb0cc4864464f9de2506724447905 : FAILURE in 2m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5e145cb98cb641128a5b7e50077a215d : FAILURE in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cca442f4209461f8cd8833d6587080a : FAILURE in 3m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a873792a076f49e4b8784d18d5fd28a5 : FAILURE in 23m 00s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/93b15ac8ab8c4e2fa73692527909045d : FAILURE in 23m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7bc5c62d26854f84e15e6be0681bb708d33f51d","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-04-27 13:46:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d9e3ed97b9c5e04adcd7b055ebf4c74788b4a02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 14:22:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28c2cc1374b94d9394af7b0a4a6ff9e2 : FAILURE in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a765242cc584fb8b76f05f622d30dd8 : FAILURE in 5m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/440a192facbe485db7a2f78443aab529 : FAILURE in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482c065b07d74ddc8dafed4d1eb5170a : FAILURE in 6m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5a7f9eeef8a24cae92f8f3e022512ecd : FAILURE in 25m 21s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/abdf5fd810cd467d92b836564c862960 : FAILURE in 28m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6e4a4bb7f05bad65f1f0b2effe08c8184773bcb","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-08-03 11:08:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d5cc2b11bebcb9371cbbe5385f96500b62795c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 14:09:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/396bf225c9594d7c85d6939f6d653a26\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9ae29ead0bc442089a9d674ee345aca : FAILURE in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9d4ffeacbc9411f9ae27e658e7ad58c : FAILURE in 3m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/eb5ecb6f00074c1a8ae8cf4dee21f796 : FAILURE in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aad508b56b0480faf2f65f4eb97a1dc : FAILURE in 2m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/324bdf2fed814658adc4be68d946e440 : FAILURE in 33m 40s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/8275da53c4e04b8faeb99bf2605bdb0b : FAILURE in 2h 55m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37e45b687a0dab2095359b77acacd3db8be8fada","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-08-08 08:39:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed72da95f71e74ff937b8794094a39f0b6b410f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 11:42:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/60b340032844444f9c4dc0149f8e53a7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0605a876d60340d4ae5fd45f8f485ada : FAILURE in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2b3084e11d440fa99f015c599aee92a : FAILURE in 3m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0fb1c8047df84a06849d71bcb1705ab8 : FAILURE in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e654e3b52db94e2fb50ffb4cea27d4f5 : FAILURE in 3m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1be8bfc864fe47f38061776950aed482 : FAILURE in 17m 02s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/0d6007b8c78d41649b1a1e45fe197e4d : FAILURE in 2h 55m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85193d3d991f7c28322e69623b310300fbf2dd38","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-08-24 09:39:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f52b0b322c850f24b2a45d1675f93d081f5ee83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 12:40:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/490fd8b3619b470c9fb6fe395c9ec4e4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e7402ebb1d641c0b27394ac6883ccd4 : FAILURE in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08e15b3e6d1a43f3bef483ad95832644 : FAILURE in 3m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f9c427dce1574884b60f717756ad21a8 : FAILURE in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc382df75074a5091c414727553d638 : FAILURE in 2m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e9c39e39ddf04048915610fc2f418cd8 : FAILURE in 23m 09s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/dfaeb9da913c42d4a1a7c2fd9481747b : FAILURE in 2h 54m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"068ed77788fcf5436e6e685bda81bbbd0205b863","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-30 16:24:32.000000000","message":"Abandoned\n\nThis branch of this project has transitioned to End of Life, open patches need to be abandoned to be able to delete the branch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0854d54ccd0f2421c6736444ab178262149c6a24","revisions":{"0854d54ccd0f2421c6736444ab178262149c6a24":{"kind":"REWORK","_number":1,"created":"2020-09-01 07:18:58.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/91/749191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/91/749191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/749191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/749191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/749191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/91/749191/1"}}},"commit":{"parents":[{"commit":"4a655241e87fc908b36fab8fa480e33406bed7bd","subject":"Use upper constraints from opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/4a655241e87fc908b36fab8fa480e33406bed7bd"}]}],"author":{"name":"Nikita Gerasimov","email":"nikita.gerasimov@oracle.com","date":"2018-11-08 16:26:03.000000000","tz":180},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-01 07:16:41.000000000","tz":120},"subject":"Fix handling of exc_info in OSJournalHandler","message":"Fix handling of exc_info in OSJournalHandler\n\nexc_info is exception tuple not intended to be printed. It\u0027s using to\ngenerate output by logging.Formatter().formatException() instead.\nexc_text used to cache this output. So we use handler formatter to get\nstring from exc_info and send it as EXCEPTION_INFO. EXCEPTION_TEXT is\nkept for backward compatibility.\n\nAlso adds type validation for journal.send fields to unit tests. Known\ntypes could be gotten from journal.DEFAULT_CONVERTERS. The rest payload\nmust be either a string or bytes.\n\nChange-Id: Idc0b9edc61b4ec39dc9df0ea8282cc104e4d83c0\nCloses-Bug: 1802142\nDepends-On: https://review.opendev.org/#/c/749189/\n(cherry picked from commit a2956d7ff08f7657dc68c78f860bcc9eb2e5a8fb)\n(cherry picked from commit 7b39c0bdf3a52d66df363de2884434204a8efe3f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0854d54ccd0f2421c6736444ab178262149c6a24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0854d54ccd0f2421c6736444ab178262149c6a24"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
