)]}'
{"id":"openstack%2Foslo.log~751476","triplet_id":"openstack%2Foslo.log~master~I7416f5611f80725280567c40a6c08f7a65df6152","project":"openstack/oslo.log","branch":"master","topic":"add-wallaby-python-jobtemplates","hashtags":[],"change_id":"I7416f5611f80725280567c40a6c08f7a65df6152","subject":"Add Python3 wallaby unit tests","status":"MERGED","created":"2020-09-11 20:58:25.000000000","updated":"2020-09-14 16:16:05.000000000","submitted":"2020-09-14 15:55:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751476-1600098950512-06ebbb73","meta_rev_id":"a7c291f5a824f8b62867f115ae1f419bcdf43a0f","_number":751476,"virtual_id_number":751476,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-14 15:55:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-11 21:03:52.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-14 08:22:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-14 12:11:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 21:03:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-14 12:11:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-14 15:55:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0a8603561834fbda4939fe55165ab3186291a5db","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-11 20:58:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a41e81c8ad18784ff63fab786adcf57ffc052cb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 21:03:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d5a396f40b920d0aba6e7d77c48dc0d5d912b160","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 22:43:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f772d11fd3545e3b7eb014ef31966ec : SUCCESS in 1h 21m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34fe0eb247a244e0bbc381457d678e98 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f730c3c0bc6b4b79852048c42c8e543f : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9aa60e0c6c33405f94cea6cacf3f6a46 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/453eafe3ba5b4614bb723dd5b1975e3a : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e4e6e2b7c9a4642bf6e4fba9ff26529 : SUCCESS in 5m 27s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/822eba566d364afaae764137401ca0f3 : SUCCESS in 59m 10s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/3ed5c0dc5d2242baac382d12d0bb1a03 : SUCCESS in 38m 37s","accounts_in_message":[],"_revision_number":1},{"id":"17bbcec508f9f5e6df3fdd783137ea50b754dec0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-14 08:22:31.000000000","message":"Patch Set 1: Code-Review+2\n\nAs we are in FF waiting a bit before merging these changes to avoid conflicts","accounts_in_message":[],"_revision_number":1},{"id":"d8c10ff9aa0f403b096202af9f1d87455fb6adcd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-14 12:11:06.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"17d0b81e693499f19acb1ab230e53c289c5aae20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 14:24:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7fe09f8344ea4ef21c6f679efb143f7e352b5a3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 15:55:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8076766fd18dbaf1541fb0235287f081e181fce4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 15:55:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2e5c2fb66b24dec8c139d6a41538b2d : SUCCESS in 1h 20m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14dac38d88994107b6978f58a9ddc24a : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f59f7aeea92a4529a4bb594d291c40d6 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93dec42189ed43c584f70e06def66332 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a39738c9f1143e4a33f4d1b37ee8524 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6db88cbea8594026a1def292b1c2233f : SUCCESS in 5m 30s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/4c3cd11f5f5447bcb5cb5bd162452039 : SUCCESS in 45m 20s","accounts_in_message":[],"_revision_number":1},{"id":"349d7866660864d7aca9fdf3beccb1c139e3f128","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 16:16:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df66fd5f1e2046e58285418413477884 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a900edfff1cb55f1447c7a987d5870f21ce5d5eb","revisions":{"a900edfff1cb55f1447c7a987d5870f21ce5d5eb":{"kind":"REWORK","_number":1,"created":"2020-09-11 20:58:25.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/76/751476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/76/751476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/76/751476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/76/751476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/76/751476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/76/751476/1"}}},"commit":{"parents":[{"commit":"31914f16aaefffeca51a67fb8848ed8d22476927","subject":"Update master for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/31914f16aaefffeca51a67fb8848ed8d22476927"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-11 20:58:23.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-11 20:58:23.000000000","tz":0},"subject":"Add Python3 wallaby unit tests","message":"Add Python3 wallaby unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for wallaby.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I7416f5611f80725280567c40a6c08f7a65df6152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a900edfff1cb55f1447c7a987d5870f21ce5d5eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a900edfff1cb55f1447c7a987d5870f21ce5d5eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
