)]}'
{"id":"openstack%2Foslo.log~772628","triplet_id":"openstack%2Foslo.log~master~I04e49f005841eb341c2d53cbc469252c089d1ab5","project":"openstack/oslo.log","branch":"master","hashtags":[],"change_id":"I04e49f005841eb341c2d53cbc469252c089d1ab5","subject":"Drop lower-constraints","status":"MERGED","created":"2021-01-27 05:55:28.000000000","updated":"2021-04-16 13:15:48.000000000","submitted":"2021-04-16 13:13:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772628","meta_rev_id":"5ceb9bae6b426cc241aadf09ed3b6e7b1f74ec7c","_number":772628,"virtual_id_number":772628,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-16 13:13:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-04-16 13:15:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-04-16 10:25:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-27 07:19:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-03 11:49:45.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-02-07 09:06:07.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"5dd77ce3b149984a2d7504411388710fb729e821","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-01-27 05:55:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eac4fc1bfe9ffa6d5de3ffb452a217762327b94e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 07:19:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/04c997eec89f4eecb971ddbd27e43c1b : SUCCESS in 2m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff9381c67b97474798af4f55cb08682a : SUCCESS in 1h 06m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ee011a044884231be349c90c175d50f : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b26e91bb8b1643069abadbeb4d25a011 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/112a845a6be948ff9290a807369f587a : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d38f5cdae4f34750a53217a2de6db4d2 : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e5cd02e833c4997a9c510dec4f295ed : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f406f1e046c459d84f3afe53a9f085d : SUCCESS in 3m 24s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/15790c013113462fa9c4c106a4fa1cbd : SUCCESS in 1h 00m 32s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/596ffdb154654571888d8c1e1826976e : SUCCESS in 38m 45s","accounts_in_message":[],"_revision_number":1},{"id":"87623f2a0c425b1c2e829db0b4f89130bad7fd45","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-03 11:49:45.000000000","message":"Patch Set 1: Code-Review+2\n\nThe commit message isn\u0027t truly correct, but this is good enough","accounts_in_message":[],"_revision_number":1},{"id":"caf73669fdda9f9f6613442061db53e49266b69c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:25:19.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ccde138031fd998dcc008b91b631e2164824d832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:25:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0fa6bdbe25b5693a444e8a7b149c0265b316db00","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:25:49.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6c8c1c50c7af45e4a65b81a77397ef7bae6fd6ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 11:46:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d75b3f71d2174ef89cec87ada8f6db22 : SUCCESS in 2m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dbd89202740043d0928273e343db57d0 : SUCCESS in 1h 19m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f30ccfae057e499ba31993239fb81b01 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cdad46030ff4f428f74591923122e59 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abdc88ec89c243c9bda88c87301cb4b1 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c02a9460b043494ebc3c7c1d5c0cd402 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b5721ec6958404fadc246d39f727d3f : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d6f48fe23ee42b7916054f0132e6b44 : SUCCESS in 3m 54s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/a410b4272e154791941503d42a227004 : SUCCESS in 1h 08m 45s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/46a7a0b76ea9481c9598027b9764f529 : SUCCESS in 40m 26s","accounts_in_message":[],"_revision_number":2},{"id":"b926386b2d3ed3f325aae0e8ee72263917f7d06d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 11:47:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d12da2d18a99fa98f828d1a75e134a1bee40bf23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:13:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/121afa3d2cae406e916f3af99033ab79 : SUCCESS in 3m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76cd42a38be74ba29dca05fa625b4b3c : SUCCESS in 1h 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38f37e52950d4629a1723d915ce710c3 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86c687c9d4294bfb91814f2e1233cb3a : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30f7aaed859c43f38c9db2e5eb1c9d22 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eee0e1af38fb40e5ab1b4f26799cd661 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe117117e2f8469cb1527086a69d58de : SUCCESS in 5m 05s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/5e123798ff5b4876adb1ae130346cf8b : SUCCESS in 36m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e7d5ebf9f653134b6ca4bedfcd3c56862001a362","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:13:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fc53581963d0e9a993d83af501903866064dea14","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:15:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd889287f4ff485ba403fdbc55ffaeb3 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53a89b9113854b98ad558fda87cb9de9 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"92f0e8f29c07751d09c96df9d502219bb5cfe9c7","revisions":{"3e564d31bb04ac313a83fc3bde7f7a120cae7a31":{"kind":"REWORK","_number":1,"created":"2021-01-27 05:55:28.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/28/772628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/28/772628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/28/772628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/28/772628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/28/772628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/28/772628/1"}}},"commit":{"parents":[{"commit":"17be566d5424ecbe95a30fcb97c615017cb9bc86","subject":"Merge \"Use py3 as the default runtime for tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/17be566d5424ecbe95a30fcb97c615017cb9bc86"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-01-27 03:23:20.000000000","tz":0},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-01-27 05:50:24.000000000","tz":0},"subject":"Drop lower-constraints","message":"Drop lower-constraints\n\nLower-constraints is not a requirement of the OpenStack Python PTI\n[0] and there currently is a discussion on the mailing list [1]\nabout dropping the test, with the oslo team already having done\nso [2].\n\nThe new dependency resolver in pip fails due to incompatible\ndependency versions in our lower-constraints file, meaning that\nwe were never providing any real guarantees with it.\n\nTo unblock the CI, I am disabling lower-constraints job for now,\nwith the option to reenable it in case we fix the constraints,\nand based on the outcome of the mailing list discussions and\nconsensus.\n\n[0]. https://governance.openstack.org/tc/reference/pti/python.html\n[1].\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-January/019672.html\n[2].\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-January/019659.html\n\nChange-Id: I04e49f005841eb341c2d53cbc469252c089d1ab5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3e564d31bb04ac313a83fc3bde7f7a120cae7a31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3e564d31bb04ac313a83fc3bde7f7a120cae7a31"}]},"branch":"refs/heads/master"},"92f0e8f29c07751d09c96df9d502219bb5cfe9c7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-16 10:25:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/28/772628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/28/772628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/28/772628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/28/772628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/28/772628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/28/772628/2"}}},"commit":{"parents":[{"commit":"127b3589233e769d3a19d2978a694f4bfcc59f1c","subject":"Merge \"Add Xena to versionutils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/127b3589233e769d3a19d2978a694f4bfcc59f1c"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-01-27 03:23:20.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-16 10:25:26.000000000","tz":0},"subject":"Drop lower-constraints","message":"Drop lower-constraints\n\nLower-constraints is not a requirement of the OpenStack Python PTI\n[0] and there currently is a discussion on the mailing list [1]\nabout dropping the test, with the oslo team already having done\nso [2].\n\nThe new dependency resolver in pip fails due to incompatible\ndependency versions in our lower-constraints file, meaning that\nwe were never providing any real guarantees with it.\n\nTo unblock the CI, I am disabling lower-constraints job for now,\nwith the option to reenable it in case we fix the constraints,\nand based on the outcome of the mailing list discussions and\nconsensus.\n\n[0]. https://governance.openstack.org/tc/reference/pti/python.html\n[1].\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-January/019672.html\n[2].\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-January/019659.html\n\nChange-Id: I04e49f005841eb341c2d53cbc469252c089d1ab5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/92f0e8f29c07751d09c96df9d502219bb5cfe9c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/92f0e8f29c07751d09c96df9d502219bb5cfe9c7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
