)]}'
{"id":"openstack%2Foslo.log~787991","triplet_id":"openstack%2Foslo.log~master~Ifd84fe4d7282e93998dc37cab4adda6d99d4adc1","project":"openstack/oslo.log","branch":"master","hashtags":[],"change_id":"Ifd84fe4d7282e93998dc37cab4adda6d99d4adc1","subject":"Remove references to \u0027sys.version_info\u0027","status":"MERGED","created":"2021-04-27 01:33:54.000000000","updated":"2021-06-01 12:22:21.000000000","submitted":"2021-06-01 12:21:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787991","meta_rev_id":"72dd80d0159c0f8eff7ce3fb26df121632c52d65","_number":787991,"virtual_id_number":787991,"owner":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 12:21:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-06-01 10:45:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-04-27 09:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-06-01 10:45:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 03:15:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-27 09:39:03.000000000","updated_by":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2021-04-27 09:39:03.000000000","updated_by":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-04-27 09:39:03.000000000","updated_by":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"47272663fa1dfc8051e8de83f69275fd77f401d9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"date":"2021-04-27 01:33:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbd4700ec636b99412b8f25ed28c424d5aad6acc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"date":"2021-04-27 01:37:23.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"88ef37c3178ef18a77be9e5148b86722339b6cae","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"date":"2021-04-27 02:11:06.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"ce7f40ea651db8e762644c6f0bac4b3a98d62b46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 03:15:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce6c996b194748428de4815d15796f8b : SUCCESS in 1h 25m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33539d1ca29e4560841d2f8e6ed79029 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44e9a3fcb1f646b0bb1e5b8a1552aea6 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ddd09b96fba489e84cb635eeb1faf20 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4434b6b8f39544bfaa69e97e156c359c : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e880aab2bd6b4e35a8c0870cb4013ccc : SUCCESS in 3m 31s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/21146202f0524f6087e3e7d0f6cd5b63 : SUCCESS in 1h 10m 28s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/8a61900ea1144629a86c16dca8d2ff43 : SUCCESS in 32m 55s","accounts_in_message":[],"_revision_number":2},{"id":"d26d9df7c8e9fa740d04f80e2b9a7c39f07e5f62","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-27 09:59:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5d81a9e8576bc77082024333731acd373e4ee3a9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-06-01 10:45:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9b29c4f799ba5a029947bccd5fbc7a1d31607e08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 10:46:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"900b04c4929c96e628abdcc3edf00026c6a02ecc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 12:21:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d81cca4d2a9444c9ed1e33342ab1e41 : SUCCESS in 1h 34m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9afb8a0face74f20bfad8f8a5a93545e : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b2854f525b346129cc4bff9148c6ab3 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3923f127aa54e228043e60a3b45149d : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aa58b6575f94c5aa01bb28eb9605475 : SUCCESS in 4m 10s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/5d6992d3462544b2abc72d2a2035b313 : SUCCESS in 40m 06s","accounts_in_message":[],"_revision_number":2},{"id":"bdba2c88257948e087e68fb304935371631c34fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 12:21:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"72dd80d0159c0f8eff7ce3fb26df121632c52d65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 12:22:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95acae2de2b2488fb072211332204c68 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"26d1c299384662a8b9bcd34641a15f9a1f1292fd","revisions":{"94d9dbf7306328fb01709e8786a609630a9b6009":{"kind":"REWORK","_number":1,"created":"2021-04-27 01:33:54.000000000","uploader":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"ref":"refs/changes/91/787991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/91/787991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/787991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/787991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/787991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/91/787991/1"}}},"commit":{"parents":[{"commit":"349d9dd9ee7253c48c1924724c55e65591c8dbd0","subject":"Merge \"Drop lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/349d9dd9ee7253c48c1924724c55e65591c8dbd0"}]}],"author":{"name":"Jorhson Deng","email":"dengzhaosen@inspur.com","date":"2021-04-27 01:33:54.000000000","tz":0},"committer":{"name":"Jorhson Deng","email":"dengzhaosen@inspur.com","date":"2021-04-27 01:33:54.000000000","tz":0},"subject":"Remove references to \u0027sys.version_info\u0027","message":"Remove references to \u0027sys.version_info\u0027\n\nWe support Python 3.6 as a minimum now, making these checks no-ops.\n\nChange-Id: Ifd84fe4d7282e93998dc37cab4adda6d99d4adc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/94d9dbf7306328fb01709e8786a609630a9b6009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/94d9dbf7306328fb01709e8786a609630a9b6009"}]},"branch":"refs/heads/master"},"26d1c299384662a8b9bcd34641a15f9a1f1292fd":{"kind":"REWORK","_number":2,"created":"2021-04-27 01:37:23.000000000","uploader":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"ref":"refs/changes/91/787991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/91/787991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/787991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/787991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/91/787991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/91/787991/2"}}},"commit":{"parents":[{"commit":"349d9dd9ee7253c48c1924724c55e65591c8dbd0","subject":"Merge \"Drop lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/349d9dd9ee7253c48c1924724c55e65591c8dbd0"}]}],"author":{"name":"Jorhson Deng","email":"dengzhaosen@inspur.com","date":"2021-04-27 01:33:54.000000000","tz":0},"committer":{"name":"Jorhson Deng","email":"dengzhaosen@inspur.com","date":"2021-04-27 01:37:10.000000000","tz":0},"subject":"Remove references to \u0027sys.version_info\u0027","message":"Remove references to \u0027sys.version_info\u0027\n\nWe support Python 3.6 as a minimum now, making these checks no-ops.\n\nChange-Id: Ifd84fe4d7282e93998dc37cab4adda6d99d4adc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/26d1c299384662a8b9bcd34641a15f9a1f1292fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/26d1c299384662a8b9bcd34641a15f9a1f1292fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
