)]}'
{"id":"openstack%2Foslo.log~827926","triplet_id":"openstack%2Foslo.log~master~If1e846c79477cb0fc06d5702240e1dbb9f2c9c40","project":"openstack/oslo.log","branch":"master","topic":"yoga","attention_set":{},"removed_from_attention_set":{"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2022-02-11 14:48:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If1e846c79477cb0fc06d5702240e1dbb9f2c9c40","subject":"Add Yoga to versionutils","status":"MERGED","created":"2022-02-04 21:33:23.000000000","updated":"2022-02-11 14:50:43.000000000","submitted":"2022-02-11 14:48:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827926-yoga","meta_rev_id":"9e2db1fd3638e0b0bc4f8eeb5b2d1300d1567048","_number":827926,"virtual_id_number":827926,"owner":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-11 14:48:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-02-07 18:45:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-10 13:13:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-10 13:13:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 22:57:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-07 18:45:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-02-10 13:13:24.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"6a13e1372abdb9a8f6fcdfa9349361b70b8d2962","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-02-04 21:33:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c07243b1f1a5783265daf8330a7a7aaeea9b511f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 22:57:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf3f08ecf68541f780dc06e79a7e8184 : SUCCESS in 1h 17m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e51b9ae4e9d840a9b335d336bf0803ab : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4541fde29edb47d0b12a76a44635841c : FAILURE in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a212c0421ef44da88c03d7916edb2157 : FAILURE in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8922452e42714241be83f53ca659be65 : FAILURE in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba39ed75edc74db28e1e389f67dac3b0 : SUCCESS in 3m 39s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/845ee9caf3db4d4da961f7c1a397391e : SUCCESS in 54m 22s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/e97baf3f71cc4ce391d3eec39c5f5395 : SUCCESS in 30m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2e377ba513fb227eafa56318d2fa4d16cbae4d30","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-02-05 19:03:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b597e652c917fe80b8245937adacbbd6bd006bb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-02-05 21:01:14.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"42b346dd45f0843d2ccdce7b20b6be7a6284cdd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 22:38:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74fce71d2ef0471aafbc0c305f7b0e26 : SUCCESS in 1h 31m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f12d7397e470475bae19a21ce9eea2a1 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d08cab39ad784a3993a8d92ef2db9e7b : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8e48fed7b214fa19d5a07f312642bd4 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45b9134814c34586a9a9a09d0afe9a28 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cfde33b7bbd4f2ba9ea6880059f9a15 : SUCCESS in 4m 04s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/6361604018d9478bb00279de5a1cf9a6 : SUCCESS in 49m 33s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/4049e0e106c04f50baa38cf1eea6a674 : SUCCESS in 28m 57s","accounts_in_message":[],"_revision_number":2},{"id":"3a7e0ba08eb86ca2423326c5a0c9070f67226847","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-07 18:45:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"00c481ef9f273a67d5ecb791a2dc403f3fe8531e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-02-10 13:13:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41ea4fbcde90b1db7b9f5aa79e13d8a2ebb4001e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 13:17:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1baad377ef62cb7cb8f3d078370e8499c760b577","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 14:48:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ddb53dbfd22b4e57b6225c452b135375 : SUCCESS in 1h 25m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ba1bbc6aa21402aa8cf7e952e137c29 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34b7531d35af4b3ab136f35f97f2eb88 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a8f34c2cb39462e9a78c281c7b4a118 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15aafc0cc1364e8e9a8704cc152041c8 : SUCCESS in 3m 39s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/44c69b40649c406ba744edc0806d678e : SUCCESS in 39m 02s","accounts_in_message":[],"_revision_number":2},{"id":"4eb7b3c8fb27b946e37c07cb4e52f71ccc727a68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 14:48:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9e2db1fd3638e0b0bc4f8eeb5b2d1300d1567048","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 14:50:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8c339e5e4df44838ade2d140870ef5a : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de03a59ed78e3e5d97d530918a4a1680160660dc","revisions":{"e0e3178dc34d8d23f65c91a2d22ac79d2cb072e8":{"kind":"REWORK","_number":1,"created":"2022-02-04 21:33:23.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/26/827926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/26/827926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/26/827926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/26/827926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/26/827926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/26/827926/1"}}},"commit":{"parents":[{"commit":"27d159b4d6364ea6284c6e1e9aa4abcda6d536fe","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/27d159b4d6364ea6284c6e1e9aa4abcda6d536fe"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2022-02-04 21:30:33.000000000","tz":-480},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2022-02-04 21:31:10.000000000","tz":-480},"subject":"Add Yoga to versionutils","message":"Add Yoga to versionutils\n\nChange-Id: If1e846c79477cb0fc06d5702240e1dbb9f2c9c40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e0e3178dc34d8d23f65c91a2d22ac79d2cb072e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e0e3178dc34d8d23f65c91a2d22ac79d2cb072e8"}]},"branch":"refs/heads/master"},"de03a59ed78e3e5d97d530918a4a1680160660dc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-05 21:01:14.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/26/827926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/26/827926/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/26/827926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/26/827926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/26/827926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/26/827926/2"}}},"commit":{"parents":[{"commit":"650ce7d313e00525c4391692fbdc4cba529e11ea","subject":"Use project when logging the user identity","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/650ce7d313e00525c4391692fbdc4cba529e11ea"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2022-02-04 21:30:33.000000000","tz":-480},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2022-02-05 21:00:13.000000000","tz":-480},"subject":"Add Yoga to versionutils","message":"Add Yoga to versionutils\n\nChange-Id: If1e846c79477cb0fc06d5702240e1dbb9f2c9c40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/de03a59ed78e3e5d97d530918a4a1680160660dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/de03a59ed78e3e5d97d530918a4a1680160660dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
