)]}'
{"id":"openstack%2Foslo.log~838185","triplet_id":"openstack%2Foslo.log~master~I7982cdb280599fb38362eb837cb527592267b91f","project":"openstack/oslo.log","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-04-25 14:12:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7982cdb280599fb38362eb837cb527592267b91f","subject":"Add system scope information to default user identity string","status":"MERGED","created":"2022-04-15 14:50:07.000000000","updated":"2022-04-25 14:14:23.000000000","submitted":"2022-04-25 14:12:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838185-secure-rbac","meta_rev_id":"6e2530949bda1163a6f449340ad5db3b91ff0e42","_number":838185,"virtual_id_number":838185,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-25 14:12:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-04-19 12:49:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-25 12:41:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-25 12:42:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-15 14:55:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-04-15 16:44:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-25 12:41:42.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"29ba3135293148b3a8ab4302e07ed280e54f061a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-15 14:50:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"501982b2dd5e49f62dd950657b845916e07118a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-15 14:51:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"76aff92903f2b72f0259e475da4aa69e44dbbf82","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-15 14:55:56.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b1491c3202ea0ecc8aa08b638d4b26e390d6a224","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 16:44:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c85ced1e13084e77a939d69349b8de96 : SUCCESS in 2m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96df9142cd714d33882d5c1693cc451b : SUCCESS in 1h 45m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd3fa3f68ccb4471bbc1fa68ffac41c7 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b04e79694204893a839cc79fb321fda : FAILURE in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b39efdce31846d8a90c447fc73da850 : FAILURE in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc567e381338430bb608bf9fc0b52812 : FAILURE in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ec8f88cd3104569a88de071b53a2508 : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/703130fec445411288c543351a12ffe4 : SUCCESS in 3m 25s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/8cc5d90d07754d209f67ab74c032ddb6 : SUCCESS in 1h 08m 47s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/adf192bddfb04102ad61ce4cc351707b : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":2},{"id":"3aefe1158c21c657a0346bcf5d8adc2107b2736f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-19 08:09:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bcb44cc405d9a6b5921ab021bc26a3743bbedb49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 09:43:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cdda5fa0e6ab4569adaae492adcdda16 : SUCCESS in 2m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56ac14684e5a468095a10df52f130838 : SUCCESS in 1h 31m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00c9ba8412404f39b28beffd9b185047 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5791f25c06ab4c13895c301211c1dedd : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5e755059ae546bfa8f6f310ab91dc42 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b26cea744377442d829b17bed8bb306d : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95c1cea8d1554396adfe0fed67c15e97 : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2b98c780a554d31a6d913d8881a6c4f : SUCCESS in 3m 52s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/b8acf351ffba461a87d86836e1bb2e5d : SUCCESS in 53m 55s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/e9d00e4fac8746e6906ea621df38a5ff : SUCCESS in 33m 44s","accounts_in_message":[],"_revision_number":3},{"id":"c4b496b7b77e974109fcaf4f80a76b5f3abef40f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-19 12:49:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"da193d0d32efea2b4f1b5a0e1db01f87f666d4cc","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-04-25 12:41:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"34d5695bb7ac3618e974b01b25e73060d93bc9d5","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-04-25 12:42:06.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ab3dfc4cb1af77379c8105c8c43d73332265147e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 12:42:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1bb14f5b0d9c513597a660239491c0f62b2f45a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:12:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f25b8b0c8ede4fcb8af6ae2e3d2be5ff : SUCCESS in 2m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b3302a95e3644aabad3a8eb7cd32808 : SUCCESS in 1h 25m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b54f41f24a4b21ad0d8a1334c7c0e8 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb7d00a9a5634cc88a3a2eb851125ed6 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9da4c71c290c46b09d5f16a3c2775794 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1fe88127c84843ae8c19259b5008b1 : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/877eb3dc154449b69b49d8d37c0eac90 : SUCCESS in 3m 27s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/8e0ce319ae8b418a9cf03255541dd3c4 : SUCCESS in 32m 27s","accounts_in_message":[],"_revision_number":3},{"id":"3485f9c1083c6da371ed82a5de0e48068c93d0d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:12:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6e2530949bda1163a6f449340ad5db3b91ff0e42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:14:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3954d96a585345108af5ccedc639eafa : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e45c71e8f77483e9f7c23564f8c7379 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7ed065f38df0595881d07eb91cec9e5c07cb38be","revisions":{"990803a05aa87c1314260b5479ae5591429344b2":{"kind":"REWORK","_number":1,"created":"2022-04-15 14:50:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/838185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/85/838185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/85/838185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/85/838185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/85/838185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/85/838185/1"}}},"commit":{"parents":[{"commit":"f31946fd8925887ae026e99a504866339c1806de","subject":"Merge \"Fix formatting of release list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f31946fd8925887ae026e99a504866339c1806de"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-15 14:47:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-15 14:47:33.000000000","tz":540},"subject":"Add system scope information to default user identity string","message":"Add system scope information to default user identity string\n\nAs part of Secure RBAC work, support for system users(eg. system admin)\nis being added. However current default format does not include\nthe system scope information so we can\u0027t determine whether the context\nis non-scoped or system-scoped.\n\nThis change updates the default logging_user_identity_format so that\nsystem scope is logged by default.\n\nChange-Id: I7982cdb280599fb38362eb837cb527592267b91f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/990803a05aa87c1314260b5479ae5591429344b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/990803a05aa87c1314260b5479ae5591429344b2"}]},"branch":"refs/heads/master"},"1f8968bbff4b61d659ccef525760f7f9fd0b8f10":{"kind":"REWORK","_number":2,"created":"2022-04-15 14:51:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/838185/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/85/838185/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/85/838185/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/85/838185/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/85/838185/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/85/838185/2"}}},"commit":{"parents":[{"commit":"f31946fd8925887ae026e99a504866339c1806de","subject":"Merge \"Fix formatting of release list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f31946fd8925887ae026e99a504866339c1806de"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-15 14:47:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-15 14:51:40.000000000","tz":540},"subject":"Add system scope information to default user identity string","message":"Add system scope information to default user identity string\n\nAs part of Secure RBAC work, support for system users(eg. system admin)\nis being added. However current default format does not include\nthe system scope information so we can\u0027t determine whether the context\nis non-scoped or system-scoped.\n\nThis change updates the default logging_user_identity_format so that\nsystem scope is logged by default.\n\nChange-Id: I7982cdb280599fb38362eb837cb527592267b91f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1f8968bbff4b61d659ccef525760f7f9fd0b8f10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1f8968bbff4b61d659ccef525760f7f9fd0b8f10"}]},"branch":"refs/heads/master"},"7ed065f38df0595881d07eb91cec9e5c07cb38be":{"kind":"REWORK","_number":3,"created":"2022-04-19 08:09:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/838185/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/85/838185/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/85/838185/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/85/838185/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/85/838185/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/85/838185/3"}}},"commit":{"parents":[{"commit":"f31946fd8925887ae026e99a504866339c1806de","subject":"Merge \"Fix formatting of release list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f31946fd8925887ae026e99a504866339c1806de"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-15 14:47:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-19 08:08:58.000000000","tz":540},"subject":"Add system scope information to default user identity string","message":"Add system scope information to default user identity string\n\nAs part of Secure RBAC work, support for system users(eg. system admin)\nis being added. However current default format does not include\nthe system scope information so we can\u0027t determine whether the context\nis non-scoped or system-scoped.\n\nThis change updates the default logging_user_identity_format so that\nsystem scope is logged by default.\n\nChange-Id: I7982cdb280599fb38362eb837cb527592267b91f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/7ed065f38df0595881d07eb91cec9e5c07cb38be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/7ed065f38df0595881d07eb91cec9e5c07cb38be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
