)]}'
{"id":"openstack%2Foslo.log~840598","triplet_id":"openstack%2Foslo.log~master~I46e2456b9931d215c97b15ac97b47a6977282584","project":"openstack/oslo.log","branch":"master","topic":"drop-py36","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-05-11 15:05:12.000000000","reason":"Hervé Beraud replied on the change"}},"hashtags":[],"change_id":"I46e2456b9931d215c97b15ac97b47a6977282584","subject":"Drop python3.6/3.7 support in testing runtime","status":"MERGED","created":"2022-05-05 09:18:12.000000000","updated":"2022-05-11 18:05:25.000000000","submitted":"2022-05-11 18:04:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840598-drop-py36","meta_rev_id":"929f67512d970b2db5071501557aa46d333bb0b9","_number":840598,"virtual_id_number":840598,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-11 18:04:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-05-05 16:57:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-05-11 14:20:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-05 22:57:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2022-05-11 14:20:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 10:49:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-05 16:57:20.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-05-05 22:57:36.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"6cbd0f7ff7a01a982372d61c193415b6f52dae8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-05 09:18:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0dee3d75ce9952a91b31e83a9625d3267b146c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 10:49:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9862e8d9431e459ebd35479caa04c116 : SUCCESS in 2m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/207dc860c7d24d66b7de704a39b6873e : SUCCESS in 1h 29m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19ecb826d418468ca4c91f534d4f4b63 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/015c2c98af664b89b5a657ae07d2c76f : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d66acf88d63a4cc9ba4d15abc37066ea : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0d5a5bc6c4a4e239e76201d61f0df99 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fa58ad02f3447cf9fdc8f18fa1b00f6 : SUCCESS in 3m 28s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/0c4cc782891447469cf04ba455bc9191 : SUCCESS in 44m 26s","accounts_in_message":[],"_revision_number":1},{"id":"64c7ecef5ffb755dea2722a23db1657d88df4229","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-05 14:13:48.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5cbfa005792e14393a537904f0fd926424e8202b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 15:36:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/04260e2cecbd43399beea9747e84f322 : SUCCESS in 2m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8b39213c9a540e79e30af8f8a8e91c0 : SUCCESS in 1h 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/129edba6b87f4781a277b23b8552a819 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/271a4c0e78804493886f752ec7ac2a04 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bae9f09b2d14451cb5eb7c7336dd86b1 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d29ffcbe75164c32a0002fe69062cf9f : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0df73a694b8f4dcfb73cdea8a866851c : SUCCESS in 3m 30s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/3968edbc53b54eda8e3f72c61a527bf6 : SUCCESS in 47m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e60eadce842b3298cd6fb1ffa670e69716e3f238","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-05 16:57:20.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bbb6ade37b820f461adce0f9fadd694a554ad38a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-05 22:57:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"961faae14cd9307b62e6684b1540bafd71166d89","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-11 14:20:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e693b86f21cb1782b6098858d06b0612e64fe43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 14:21:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14fad640b9c2f61619bb14b6aad17d0f39853fac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 14:53:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/75fca517a14d4788bd145d7012579b73 : SUCCESS in 2m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4a97200bc394507aefb3f6104a0de6d : FAILURE in 20m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb5e1e4b2c794afda0066755f957d30f : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b3bdc799b934be091a3dfe7308b6466 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfb303b6af0a451eb735a8adae6af1fa : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91e3273f15274630876ef4c53f2c9b0f : SUCCESS in 3m 25s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/3fdd85115b79413fbb6234f2553c5a81 : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":2},{"id":"a33dda33b97e331d119e2ee9182a08b519f6b6d5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-11 15:05:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"862b11a717e2ad40ecb562796f06a05fa0874762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 16:28:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/da19e1b67b8d478f8ae52b0542b83d92 : SUCCESS in 3m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7510de2881a84b2482feec9d5bb5546d : SUCCESS in 1h 20m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77f3f022937b4293be590dbe72a3c416 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3a3d25fcd224bba9a32a68d824a7afb : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45c5e6a0ae6d4451b5f4e9a9d4cceb9d : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63edf5a97a084832bf07a37bb290f225 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1368f6d8c1d14e0ea41af22bb9106b89 : SUCCESS in 3m 47s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/90d97ae0027e47798c1da82f28d3e5ed : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":2},{"id":"92bff6a0fecc2c40e93abf01694050c0cb21c46f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 16:28:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dd168bed8c1a63553e32a7bfd55158e96f110ff9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 18:04:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f2976aa21a8248c3831c9321aa855fcf : SUCCESS in 2m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/263d0e1a0c69454483b5ff297e67bd25 : SUCCESS in 1h 35m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89f16f696c424cd2a9cb0b82fa9da429 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abdc9137e6064853831c7072ff259e36 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7741f915bc9400a87090bf5330f8829 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d831f3616f14c629d38e959002596a8 : SUCCESS in 3m 05s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/0c3468c98bfc43e58e069ccc2f118df0 : SUCCESS in 44m 53s","accounts_in_message":[],"_revision_number":2},{"id":"8cc20c87524148046c00f8a3b83216e18ba8e24b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 18:04:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"929f67512d970b2db5071501557aa46d333bb0b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 18:05:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6fdc66aa26f4f358bba2aa5e5448ff1 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cdbf6e7a915059df7820b41a989adce09969834c","revisions":{"1a331ea292440a519ef8344c3a641b7b5db4b969":{"kind":"REWORK","_number":1,"created":"2022-05-05 09:18:12.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/98/840598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/98/840598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/98/840598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/98/840598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/98/840598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/98/840598/1"}}},"commit":{"parents":[{"commit":"ebdee7f39920ad5b4268ee296952432b0d41a375","subject":"Merge \"Log Global Request IDs by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ebdee7f39920ad5b4268ee296952432b0d41a375"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 09:17:59.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 09:17:59.000000000","tz":120},"subject":"Drop python3.6 support in testing runtime","message":"Drop python3.6 support in testing runtime\n\nIn Zed cycle testing runtime, we are targetting to drop the\npython 3.6 support, project started adding python 3.8 as minimum,\n\nexample nova:\n- https://github.com/openstack/nova/blob/56b5aed08c6a3ed81b78dc216f0165ebfe3c3350/setup.cfg#L13\n\nChange-Id: I46e2456b9931d215c97b15ac97b47a6977282584\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1a331ea292440a519ef8344c3a641b7b5db4b969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1a331ea292440a519ef8344c3a641b7b5db4b969"}]},"branch":"refs/heads/master"},"cdbf6e7a915059df7820b41a989adce09969834c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-05 14:13:48.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/98/840598/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/98/840598/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/98/840598/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/98/840598/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/98/840598/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/98/840598/2"}}},"commit":{"parents":[{"commit":"ebdee7f39920ad5b4268ee296952432b0d41a375","subject":"Merge \"Log Global Request IDs by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ebdee7f39920ad5b4268ee296952432b0d41a375"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 09:17:59.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 14:13:06.000000000","tz":120},"subject":"Drop python3.6/3.7 support in testing runtime","message":"Drop python3.6/3.7 support in testing runtime\n\nIn Zed cycle testing runtime, we are targetting to drop the\npython 3.6/3.7 support, project started adding python 3.8 as minimum,\n\nexample nova:\n- https://github.com/openstack/nova/blob/56b5aed08c6a3ed81b78dc216f0165ebfe3c3350/setup.cfg#L13\n\nChange-Id: I46e2456b9931d215c97b15ac97b47a6977282584\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/cdbf6e7a915059df7820b41a989adce09969834c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/cdbf6e7a915059df7820b41a989adce09969834c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
