)]}'
{"id":"openstack%2Foslo.log~852443","triplet_id":"openstack%2Foslo.log~master~Iac1b0891ae584ce4b95964e6cdc0ff2483a4e57d","project":"openstack/oslo.log","branch":"master","topic":"eventlet-gd","attention_set":{},"removed_from_attention_set":{"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-08-15 10:12:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iac1b0891ae584ce4b95964e6cdc0ff2483a4e57d","subject":"Fix logging in eventlet native threads","status":"MERGED","created":"2022-08-08 15:10:34.000000000","updated":"2024-03-07 11:04:40.000000000","submitted":"2022-08-15 10:12:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852443-eventlet-gd","meta_rev_id":"dc66e8ab2232ef47a977d945f0320360bfcb330a","_number":852443,"virtual_id_number":852443,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-15 10:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-08-09 10:09:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-08-15 08:47:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2022-08-15 08:47:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2022-08-15 08:47:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-08 16:52:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-09 10:09:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-08-15 08:47:08.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-03-07 11:04:40.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"state":"CC"},{"updated":"2024-03-07 11:04:40.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"}],"messages":[{"id":"8094f343a97d2299921b2f96e08bcb90aa37df80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-08 15:10:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6af64d8bb14cac0e5f98ab368a4b2749a50c564a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 16:52:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff1ecd12f5874abab14b5789b833ca46\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/671a5bc2673441fc9fb9a0a6b3cd81a3 : SUCCESS in 2m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43635a51ddc64a5e8c675acb740d9377 : SUCCESS in 1h 36m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a672fb5ae574459c8be2aac53929043f : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c98eda854cc407fbc0df03ee72c9bb4 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68e82411afd54a689ef408b9291c95bc : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac8054ca973345bdbf39fbba4120de84 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3677b82ff76c49aca7b21db7ad250dc8 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1278f4e2f991451fa37311841ded453d : SUCCESS in 3m 50s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/6da2558a0d3e4a7abc838879fc287215 : FAILURE in 55m 10s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/a6360c73b8b240c58259287be7cb8aaf : SUCCESS in 40m 46s","accounts_in_message":[],"_revision_number":1},{"id":"507f52d241dcaa1930afb21088d643cd04189a8c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-08-09 10:09:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7017689b76860235134a729d82f906bbf30002a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 08:47:08.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e5643e0bc542b491f6d867ac6ec48c7bb338eb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 08:47:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8209333905d7b414d30198d438a34a504b0b2635","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 10:12:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df70e695339d4325b6889df5fa6e469f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a9818a161e244f22b6cd7434f9f7823b : SUCCESS in 2m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c62ab0f77c7b4aec87591497efd34680 : SUCCESS in 1h 24m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a7de4f72c16480c99275bd2f7c49a0f : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2ab573d62f24c9889c0ff2fe90a9106 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d97304145924599b18d61922243db1b : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/966c164b41274c1cb4e0712daeaeac8a : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eedfe5e78896447e9b428190ec1bb746 : SUCCESS in 3m 43s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/32a895584e704574912bf0a48331dfc7 : SUCCESS in 31m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f3724734f1ecf61ba3c0e08a695f7f249113a313","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 10:12:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9427fc3980d1d9831979a2a1b7bd1de2d71f3a6e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 10:13:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bed6dedad4942b6ad499a64bba2c143\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57eb1b30c50c4e1e9dc01194c6fbf5e5 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e194fb0331d4d7096fac680d8b1cb22 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"94b9dc32ec1f52a582adbd97fe2847f7c87d6c17","revisions":{"94b9dc32ec1f52a582adbd97fe2847f7c87d6c17":{"kind":"REWORK","_number":1,"created":"2022-08-08 15:10:34.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/43/852443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/43/852443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/43/852443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/43/852443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/43/852443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/43/852443/1"}}},"commit":{"parents":[{"commit":"6401da71b3b1b8d0fc213f0a58dbf69031acba87","subject":"Merge \"[Fix] init global_request_id if not in context\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6401da71b3b1b8d0fc213f0a58dbf69031acba87"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-08 14:44:20.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-08 15:09:50.000000000","tz":120},"subject":"Fix logging in eventlet native threads","message":"Fix logging in eventlet native threads\n\nThere is a bug in eventlet where logging within a native thread can lead\nto a deadlock situation: https://github.com/eventlet/eventlet/issues/432\n\nWhen encountered with this issue some projects in OpenStack using\noslo.log, eg. Cinder, resolve them by removing any logging withing\nnative threads.\n\nThere is actually a better approach. The Swift team came up with a\nsolution a long time ago [1], and in this patch that fix is included as\npart of the setup method, but will only be run if the eventlet library\nhas already been loaded.\n\nThis patch adds the eventlet library as a testing dependency for the\nPipeMutext unit tests.\n\n[1]: https://opendev.org/openstack/swift/commit/69c715c505cf9e5df29dc1dff2fa1a4847471cb6\n\nCloses-Bug: #1983863\nChange-Id: Iac1b0891ae584ce4b95964e6cdc0ff2483a4e57d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/94b9dc32ec1f52a582adbd97fe2847f7c87d6c17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/94b9dc32ec1f52a582adbd97fe2847f7c87d6c17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
