)]}'
{"id":"openstack%2Foslo.log~864252","triplet_id":"openstack%2Foslo.log~master~I4bbcfe7db6d75188e61b9084cb02b2dd2aaa0c76","project":"openstack/oslo.log","branch":"master","topic":"bug/1995514","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-11-14 20:05:21.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-11-14 18:11:33.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-11-14 20:05:21.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2022-11-14 20:05:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4bbcfe7db6d75188e61b9084cb02b2dd2aaa0c76","subject":"Make the eventlet logging fix execution conditional","status":"MERGED","created":"2022-11-11 09:27:00.000000000","updated":"2022-11-14 20:06:25.000000000","submitted":"2022-11-14 20:05:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864252-bug/1995514","meta_rev_id":"6d17048ad57ac6a6157266d5a7796f320f6a9fab","_number":864252,"virtual_id_number":864252,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-14 20:05:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-11-14 18:11:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-11-14 18:11:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-11 09:37:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-11-11 09:37:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-11 09:37:48.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2022-11-11 10:59:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-11 11:25:30.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-11-11 13:43:33.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"640731971f05758446601826687c9cf28d424222","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-11 09:27:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a9bf2e6a48ba742749c6c7c1330161475ab4a76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 10:59:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e48f198e950c4e038aed5c300d97582b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd21aab2ff0141b0b47ac9e29c84f203 : SUCCESS in 1h 29m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a13d194c8e514155beeacf3bbe8e058b : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f623124557e449e945db4e5783b5128 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25f7e84c34864faeb87787424c53baef : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36c0cee8e8884d0ca4695adc7db88296 : SUCCESS in 4m 51s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/d5ac2e88d79e4b6b9842d837950e9573 : FAILURE in 50m 47s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/48bd4ac50dd2404981c1f95dcfe6b447 : SUCCESS in 30m 36s","accounts_in_message":[],"_revision_number":1},{"id":"17bc251f111ad808cfb881f54659f183cea80d94","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-11-11 11:25:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5ed75cff1ebade23f00448eca425c7e844918101","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-11 13:43:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8efa0c0176acea2482921c97991a53267ecdf918","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-11-14 14:06:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8504d8f99bc522a31891dfcff7de7f5b4cb16fd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-14 16:58:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3a9060fbd11fa81b68cd1fa4fd91bdc51fca955","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-14 16:59:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e488e9253b916cd80643c51cabda3f307f47a378","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-11-14 18:11:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"28116ce6afa20d3d294a96bafff8b07cb41a4cc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 18:21:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcc5f8791b1e422fac5aef01f071145c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7be062dfee94e0fbaf231eff6a91d5a : SUCCESS in 1h 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e00bfbecf1b4018a5aae9b0b70931d8 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6314a332eec147f58cb465517d6c8174 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba9ac67f2cb64c1ea5cc11425228730c : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f51e7b84d27c4869a1d68318691bd0b0 : SUCCESS in 3m 38s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/00d3156445bb4d208ff3f2a33811ab58 : FAILURE in 1h 00m 38s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/0a4a24e733f74338b625f4a327ccc6b2 : SUCCESS in 40m 08s","accounts_in_message":[],"_revision_number":2},{"id":"baecff285fa83f3cb40c6b5d80eab1c831b81d20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 18:22:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"47f29e91e2f0f311e4fa86186f96d2edc6cbced0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 20:05:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c622593ada948dc8e4657e86be2752a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8a0a99d30944eceb7ff2916f8e5747b : SUCCESS in 1h 38m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f32c01a503594dee830738d3f53d1e80 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/961649fd4bcc40e3b6f0e0ca312a698d : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41d4452c0fe84b8aae893a36e86ded64 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c62b82e030914c1f939c87a767576664 : SUCCESS in 3m 00s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/715b85725b4d4a3987d5811e956ec9c3 : SUCCESS in 43m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4bb9e69b5f9fdca9b9d7d1859b03b30d8ac3fd12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 20:05:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6d17048ad57ac6a6157266d5a7796f320f6a9fab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 20:06:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd89f3a3f53e4b4aaa8f7a8c1a15054b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1627084a47241b08b705b6961916d0b : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de615d9370681a2834cebe88acfa81b919da340c","revisions":{"f0677c72ff308c6f9222704f88311965712fbcd9":{"kind":"REWORK","_number":1,"created":"2022-11-11 09:27:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/864252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/52/864252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/52/864252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/52/864252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/52/864252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/52/864252/1"}}},"commit":{"parents":[{"commit":"94b9dc32ec1f52a582adbd97fe2847f7c87d6c17","subject":"Fix logging in eventlet native threads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/94b9dc32ec1f52a582adbd97fe2847f7c87d6c17"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-11 09:15:58.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-11 09:15:58.000000000","tz":60},"subject":"Make the eventlet logging fix execution conditional","message":"Make the eventlet logging fix execution conditional\n\nThis patch changes the ``log.setup()`` signature to allow a kwarg\nthat makes the eventlet logging fix conditional.\n\nSome libraries that import oslo.log, replace the logging locking\nhandling with its own implementation. For example oslo.privsep\nreplaces the logging handler with its own one [1], that allows the\ncommunication between the privileged daemon process and the service\nimporting the library. This local handler clashes with the\nevenlet logging fix implemented in [2].\n\n[1]https://github.com/openstack/oslo.privsep/blob/1020c6fbf54f93cad9fa36b97bcd9cf75549984a/oslo_privsep/daemon.py#L579\n[2]https://review.opendev.org/c/openstack/oslo.log/+/852443\n\nPartial-Bug: #1995514\nRelated-Bug: #1995091\nChange-Id: I4bbcfe7db6d75188e61b9084cb02b2dd2aaa0c76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f0677c72ff308c6f9222704f88311965712fbcd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f0677c72ff308c6f9222704f88311965712fbcd9"}]},"branch":"refs/heads/master"},"de615d9370681a2834cebe88acfa81b919da340c":{"kind":"REWORK","_number":2,"created":"2022-11-14 16:59:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/864252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/52/864252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/52/864252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/52/864252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/52/864252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/52/864252/2"}}},"commit":{"parents":[{"commit":"94b9dc32ec1f52a582adbd97fe2847f7c87d6c17","subject":"Fix logging in eventlet native threads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/94b9dc32ec1f52a582adbd97fe2847f7c87d6c17"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-11 09:15:58.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-11-14 16:59:12.000000000","tz":0},"subject":"Make the eventlet logging fix execution conditional","message":"Make the eventlet logging fix execution conditional\n\nThis patch changes the ``log.setup()`` signature to allow a kwarg\nthat makes the eventlet logging fix conditional.\n\nSome libraries that import oslo.log, replace the logging locking\nhandling with its own implementation. For example oslo.privsep\nreplaces the logging handler with its own one [1], that allows the\ncommunication between the privileged daemon process and the service\nimporting the library. This local handler clashes with the\nevenlet logging fix implemented in [2].\n\n[1]https://github.com/openstack/oslo.privsep/blob/1020c6fbf54f93cad9fa36b97bcd9cf75549984a/oslo_privsep/daemon.py#L579\n[2]https://review.opendev.org/c/openstack/oslo.log/+/852443\n\nPartial-Bug: #1995514\nRelated-Bug: #1995091\nChange-Id: I4bbcfe7db6d75188e61b9084cb02b2dd2aaa0c76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/de615d9370681a2834cebe88acfa81b919da340c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/de615d9370681a2834cebe88acfa81b919da340c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
