)]}'
{"id":"openstack%2Foslo.log~868937","triplet_id":"openstack%2Foslo.log~master~Ic3542205ffc30924e878b103d7f4ac7c6345f33c","project":"openstack/oslo.log","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2023-02-27 12:04:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic3542205ffc30924e878b103d7f4ac7c6345f33c","subject":"Fix tox.ini for tox4","status":"ABANDONED","created":"2022-12-31 03:04:52.000000000","updated":"2023-02-27 12:04:23.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"d1c6f2d8638d38b6022dd79c75eb50a416e1e1a1","_number":868937,"virtual_id_number":868937,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-03 13:54:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":-1,"date":"2023-01-04 04:01:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-31 04:18:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-03 11:37:32.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2023-01-04 04:01:17.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"2bd06861ae02f2f47ca33eddefd387ebed3b36b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-12-31 03:04:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6dd438b46f3c4925b4da2fad01eb4dee7e83fba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-31 04:18:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d04210158ca4ba2bbf4524dcfd483ac\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac08c27177154836b98d4e878727fef5 : FAILURE in 1h 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96f849a75dbd44e1bc3682d04633f7e0 : TIMED_OUT in 31m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62dbdc45fcbc4fe48aec9b1efcd6f3e8 : TIMED_OUT in 40m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75dc8e494f4a47a2b62fe6b48bf8eaba : TIMED_OUT in 40m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ec6974654b406bb23723795ae0a59c : TIMED_OUT in 31m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0004e4754f6435ebd88bda8fda19244 : SUCCESS in 2m 49s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/0cdf306cd6044a3fbac1fcde8e0568ee : SUCCESS in 1h 07m 39s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/6ed694c3279d4d01a3be3e0910caddb6 : SUCCESS in 28m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a8dcb3dea68dd924392da726808a9230e8d1d607","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-12-31 06:29:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c383ce4871d2df50e9f757e21c678419f427ea06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-31 08:08:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/834d317603624f068592de07806a04e8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d995137b55974efaacc8f7e4b3b6dc07 : SUCCESS in 1h 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ac963a9a6f542219b72d8540feca5d8 : TIMED_OUT in 30m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e4528a5be4e457e91dcf9c5760ab080 : TIMED_OUT in 40m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de6b439a3244412f9aa68ca93e74a546 : TIMED_OUT in 41m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffbd69629aee44a482a439ec10ca1dab : TIMED_OUT in 30m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a2ed04ad57b41df8461ee8db3cf62b2 : SUCCESS in 2m 52s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/61dd0eb2c4264ec6b052ca8e4796f8ab : SUCCESS in 1h 02m 40s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/d101575e787d4ba787878b9724699827 : SUCCESS in 27m 28s","accounts_in_message":[],"_revision_number":1},{"id":"423f723200ba707b98c004bc175a43070631e5bc","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-01-03 09:23:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0f7bc8d1ad956d482d2793857cb6e042a071566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 10:48:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdf7d22e313a4a76898e626256e7c82b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fee31f55041e45428b09b26a024671b1 : SUCCESS in 1h 22m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e66357ec939e43d38a98754bcb2765dc : TIMED_OUT in 30m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a641861f9f1b4565b21d397ecea3c207 : TIMED_OUT in 40m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ded558f36ff4c759d2c80f32b7ea3e7 : TIMED_OUT in 40m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d2154b993434fc2b8d863c09a06e51b : TIMED_OUT in 31m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29ee2076864f419ab6fc8aa91344d91c : SUCCESS in 2m 50s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/00f3be46922541a7970c8721d390ac5e : SUCCESS in 1h 01m 32s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/e50cb64138134764a76e36bb36164323 : SUCCESS in 41m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b6015423c478f8707729dad2d5ffd6a9cab35845","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2023-01-03 11:37:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f20a487f39bb2d7f9aebd580280d2aa220ab2aec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-01-03 11:46:38.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"58ffed094412a517b623728362256f3d8ce40bc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-01-03 12:11:05.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3a20c76163ac3f623f1543a77e88b13ebae3ee3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 13:54:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b989a2d6a1804d5f806aa1d465ef7f58\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d79f19338df461abdcbb5af525f703b : SUCCESS in 1h 38m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35000fab9cff41d790d817143cb3bc8c : TIMED_OUT in 30m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a496467912bc4509b5eccc56fd89dab3 : TIMED_OUT in 40m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/144fcba64108499f9a69e1d472b2c879 : TIMED_OUT in 40m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ea677e84264cbcaaeae90d01b6f999 : TIMED_OUT in 30m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98c915c5fd8e43e78b341f23701919cc : SUCCESS in 2m 54s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/9ea345703dd94b38bc3c886dea72ca85 : SUCCESS in 1h 10m 32s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/f3202e3c4eef4b9cb536d6d4d3e053eb : SUCCESS in 24m 03s","accounts_in_message":[],"_revision_number":3},{"id":"f2068419573afd893b2e23694347625b512dcddd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-04 04:01:17.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d1c6f2d8638d38b6022dd79c75eb50a416e1e1a1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-27 12:04:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4285b132a8afe95d13c43a969f28f41839bbc334","revisions":{"c1592ae764377514c188efd7f093dcef79177d92":{"kind":"REWORK","_number":1,"created":"2022-12-31 03:04:52.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/37/868937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/37/868937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/37/868937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/37/868937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/37/868937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/37/868937/1"}}},"commit":{"parents":[{"commit":"de615d9370681a2834cebe88acfa81b919da340c","subject":"Make the eventlet logging fix execution conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/de615d9370681a2834cebe88acfa81b919da340c"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-12-31 03:04:17.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-12-31 03:04:20.000000000","tz":480},"subject":"Fix tox.ini for tox4","message":"Fix tox.ini for tox4\n\nUse allowlist_externals instead of whitelist_externals.\nExecute shell script with bash.\n\nChange-Id: Ic3542205ffc30924e878b103d7f4ac7c6345f33c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/c1592ae764377514c188efd7f093dcef79177d92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/c1592ae764377514c188efd7f093dcef79177d92"}]},"branch":"refs/heads/master"},"1c0fa68d15c3e17cb78294cab793c0984c1d2f5c":{"kind":"REWORK","_number":2,"created":"2023-01-03 11:46:38.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/37/868937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/37/868937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/37/868937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/37/868937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/37/868937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/37/868937/2"}}},"commit":{"parents":[{"commit":"de615d9370681a2834cebe88acfa81b919da340c","subject":"Make the eventlet logging fix execution conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/de615d9370681a2834cebe88acfa81b919da340c"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-12-31 03:04:17.000000000","tz":480},"committer":{"name":"Wenping Song","email":"songwenping@inspur.com","date":"2023-01-03 11:46:38.000000000","tz":0},"subject":"Fix tox.ini for tox4","message":"Fix tox.ini for tox4\n\nUse allowlist_externals instead of whitelist_externals.\nExecute shell script with bash.\n\nChange-Id: Ic3542205ffc30924e878b103d7f4ac7c6345f33c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1c0fa68d15c3e17cb78294cab793c0984c1d2f5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1c0fa68d15c3e17cb78294cab793c0984c1d2f5c"}]},"branch":"refs/heads/master"},"4285b132a8afe95d13c43a969f28f41839bbc334":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-01-03 12:11:05.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/37/868937/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/37/868937/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/37/868937/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/37/868937/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/37/868937/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/37/868937/3"}}},"commit":{"parents":[{"commit":"de615d9370681a2834cebe88acfa81b919da340c","subject":"Make the eventlet logging fix execution conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/de615d9370681a2834cebe88acfa81b919da340c"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-12-31 03:04:17.000000000","tz":480},"committer":{"name":"Wenping Song","email":"songwenping@inspur.com","date":"2023-01-03 12:11:05.000000000","tz":0},"subject":"Fix tox.ini for tox4","message":"Fix tox.ini for tox4\n\nUse allowlist_externals instead of whitelist_externals.\nExecute shell script with bash.\nMade 4.0.0 the tox minversion\n\nThis patch makes tox 4 the default so that we can hopefully catch\nproblems locally before they block the gate.\n\nChange-Id: Ic3542205ffc30924e878b103d7f4ac7c6345f33c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/4285b132a8afe95d13c43a969f28f41839bbc334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/4285b132a8afe95d13c43a969f28f41839bbc334"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
