)]}'
{"id":"openstack%2Foslo.log~868938","triplet_id":"openstack%2Foslo.log~master~I82753bf268ebcc6fd701a56c64534a9cd3d5b4ae","project":"openstack/oslo.log","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2023-02-17 14:57:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I82753bf268ebcc6fd701a56c64534a9cd3d5b4ae","subject":"Cleanup py27 support","status":"MERGED","created":"2022-12-31 03:13:06.000000000","updated":"2023-02-17 14:58:56.000000000","submitted":"2023-02-17 14:57:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868938","meta_rev_id":"b92768cbe0f7c9e06ce9000b6737bf91eb2959cc","_number":868938,"virtual_id_number":868938,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 14:57:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-02-17 12:07:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-02-17 12:07:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-31 05:03:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 12:07:55.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"6919f7cd0e820bb6ace6e0ab9e83b25e14b539b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-12-31 03:13:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"574bdd0e062ecbb89aaf99f52e08c86882ab7695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-31 05:03:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df37e996a8f54c32b304542552b94b82\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/54590b371ae9406f92dd9d718ce17d9a : SUCCESS in 3m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1232eaa157a14709a4bf25187ef8cf05 : SUCCESS in 1h 44m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a0e3dbe4a644e66995c6b1c063a7428 : TIMED_OUT in 30m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f8af5c22a9f45c5ba599082ce05ab5c : TIMED_OUT in 40m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a89594e9ca242779021a00079427052 : TIMED_OUT in 40m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fba86cbae4cd49739b0d42c5184ae569 : TIMED_OUT in 30m 58s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/1ffdae8a39f14fcfa39294ddae56efe0 : SUCCESS in 58m 21s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/d64ed7ba5fa445f793a66033dee9d33a : SUCCESS in 40m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f66ccf2528400ea7649cf40cf6f26466f92671b3","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-12-31 06:30:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0b7f888f5362e3740e0edf07640f27916d0a67d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-31 08:20:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/069605d75d8e475eae94aff3ae970fec\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b8faeb126f874825afcf1db89d24b6d1 : SUCCESS in 1m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14b453879c364126a4046a1f85f69cd7 : SUCCESS in 1h 21m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e01624c340445dfab49cce3450ee115 : TIMED_OUT in 30m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92be1351655949528d508110a34950f2 : TIMED_OUT in 40m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c81009a7b5cb4a44b76fc0d3ab84c95d : TIMED_OUT in 40m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22e79e0289194467992b438f10eae033 : TIMED_OUT in 30m 51s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/719ffddf126842aba1f23862a4593457 : FAILURE in 13m 35s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/2e0b66a1cdaf423ab960fc40aeebca98 : SUCCESS in 22m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9e4080d6794ff1bbdb4e06f84b5b6b68a16461c9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-17 12:07:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cab3ad2a751713174b65f2bdfd861ad8a3952bf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 13:40:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7699be7529b64e85aff28c797fa75875\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/04147060e4d3435597861a156977eab4 : SUCCESS in 2m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8fd9223659541bea27de47d05a2da95 : SUCCESS in 1h 24m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50efc7d26dc04536a7b956cf6b824d37 : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ba37243045243359c02184913d9690d : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a7ff6c969384cb78d0acc8e16d43849 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae51d72df6a24926862d3322e2bebce3 : SUCCESS in 3m 05s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/747b06085a214c1b9bcd2670a696733b : SUCCESS in 1h 03m 12s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/71c84e3faf7648bfbc7ab8da08b19cbf : SUCCESS in 30m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e9db14ca0bf549e5401a4d7e6c975b9405a58845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 13:41:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57925caa82ae21cec1b27fd31bea35b3813da44d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 14:57:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09b23debcdb84b0e90a4dfaf15a304ad\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7a03f24bf757405d9eb602e5a5e392f1 : SUCCESS in 1m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef98bba1f134457ba12df403a2afd01b : SUCCESS in 1h 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/727e85af0e714c87a59d561ff2181ebf : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec7cbf5bccc641eb9497691f8d4372bc : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea7399a687664ee793df8088ae8f347a : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e5ba121fbb041bc8843804669900399 : SUCCESS in 3m 23s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/5f889951a69242c4835eb5d44b1230f9 : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":1},{"id":"05928885738bc4121546de1d3afaa86c555b42a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 14:57:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b92768cbe0f7c9e06ce9000b6737bf91eb2959cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 14:58:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/340b8d5d022249fda8d5ee37f1129bbe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba0c89ac96b4fc2a39eb5fd195a267f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b6f58019c45b6883a33e2531744a4235d7a48b9","revisions":{"2b6f58019c45b6883a33e2531744a4235d7a48b9":{"kind":"REWORK","_number":1,"created":"2022-12-31 03:13:06.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/38/868938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/38/868938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/868938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/868938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/868938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/38/868938/1"}}},"commit":{"parents":[{"commit":"de615d9370681a2834cebe88acfa81b919da340c","subject":"Make the eventlet logging fix execution conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/de615d9370681a2834cebe88acfa81b919da340c"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-12-31 03:12:45.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-12-31 03:12:48.000000000","tz":480},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n\nChange-Id: I82753bf268ebcc6fd701a56c64534a9cd3d5b4ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2b6f58019c45b6883a33e2531744a4235d7a48b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2b6f58019c45b6883a33e2531744a4235d7a48b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
