)]}'
{"id":"openstack%2Foslo.log~877816","triplet_id":"openstack%2Foslo.log~stable%2F2023.1~I80ea7da0fe332b7fed36bbf28c0edf280ff560e2","project":"openstack/oslo.log","branch":"stable/2023.1","topic":"revert-oslo-5.1.0-functional-content","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-03-20 15:52:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I80ea7da0fe332b7fed36bbf28c0edf280ff560e2","subject":"Revert \"Make the eventlet logging fix execution conditional\"","status":"MERGED","created":"2023-03-17 15:50:55.000000000","updated":"2023-03-20 15:53:30.000000000","submitted":"2023-03-20 15:52:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877816-revert-oslo-5.1.0-functional-content","meta_rev_id":"1f1ef5fd40e1d7d92791c718a0c8b1a7ff4d856e","_number":877816,"virtual_id_number":877816,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-20 15:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-03-20 11:03:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-03-20 09:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-03-20 14:49:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-17 17:57:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-20 09:45:59.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-03-20 11:03:26.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"ecf18f254c8a88bb382f006a91b1cc9affa868a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-17 15:50:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df2035e9041a994dbd2dd8c8674590a981398bd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 17:57:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b2e8856ba724a77ae881a74dc635b85\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f7cd06a7e2b4c038a84cef249006c73 : TIMED_OUT in 2h 04m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a9d25b00404e96853fc6a0af77fa87 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/765ace2c57b342858f98d04647f44312 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92bbdb2d776c45c1b532ed2495be5492 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/302243c0f1e643a394a1a18324129b00 : SUCCESS in 3m 36s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/00ac9c7e1d1241f49fa4d3ca059cdaca : SUCCESS in 48m 30s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/b6a9caf39d954c5c8d8655487a4d2762 : SUCCESS in 39m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8c0c047c31a47d4ea8361855b80237bf60b3a44c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-20 09:45:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39f386fe0277e679243d1c4df6f769f4336acfee","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-20 09:45:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3cc3fe7c26d4474228c7ab25b6bda35e17a163f1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-03-20 11:03:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a234d093fe12a6c361c95dcb53635092407ffa80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 11:13:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10ff95d03ab34a748ef7757ba600068a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4fb63860ffa4a1297bd48d7fc739099 : SUCCESS in 1h 21m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b699c63e5414ebc92c62fa85c97329c : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bde663dfd7cb4154aa615ddae1e18552 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68f3b78afda0447dbaa09244f3cf49a3 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcce29c7486447838052836432e1fa8c : SUCCESS in 3m 47s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/5dbe4906f6474a5db783b3e27deb558d : SUCCESS in 56m 55s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/be33835ffcce494e81478a9378cfdec7 : SUCCESS in 37m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d5a6595e25ed685daf3e9b1756cd63147ed2fee8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-20 14:49:43.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b881c387805b1d25a47f35e4523d0a48160e0d70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 14:50:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b01141b177d4f4ba75e4c04fd062b8ca473edd62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 15:52:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74a4c04eb6dd4ed4beb9497220ba7ee4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1954651c339c42a283e1bc16421e76aa : SUCCESS in 1h 01m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb95f06fc9374fa89d1dcafd088ee958 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/775c8592b9e04e4bab7aa5cc8f29c606 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0105a4794df44b6caa5dd152840519c1 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d23c455043364e00893bac923834c990 : SUCCESS in 2m 58s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/314521d228e948979c49bf2ebdcb703c : SUCCESS in 45m 49s","accounts_in_message":[],"_revision_number":1},{"id":"28e518a1723fee9dca627b96bdb7e1835515381a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 15:52:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1f1ef5fd40e1d7d92791c718a0c8b1a7ff4d856e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 15:53:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b250c711ceb408bae5034d423da19f9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/958e8deb0eb44d33a0c5215fedcdf640 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"974766a26e29aa91a47edafc6ac1531f94650db2","revisions":{"974766a26e29aa91a47edafc6ac1531f94650db2":{"kind":"REWORK","_number":1,"created":"2023-03-17 15:50:55.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/16/877816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/16/877816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/877816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/877816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/877816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/16/877816/1"}}},"commit":{"parents":[{"commit":"94c11c63826d5fe1607eccd569f016297a8c88e6","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/94c11c63826d5fe1607eccd569f016297a8c88e6"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-03-17 15:24:39.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-03-17 15:33:12.000000000","tz":60},"subject":"Revert \"Make the eventlet logging fix execution conditional\"","message":"Revert \"Make the eventlet logging fix execution conditional\"\n\nThis reverts commit de615d9370681a2834cebe88acfa81b919da340c.\n\nDuring 2023.1 Antelope cycle the upper constraint version bump patch\ncould not merge [1], as cross-nova-py310 fails. This means that during\nthe cycle everything was tested with oslo.log 5.0.0, so as we discussed\nthis during release meeting [2] we need to revert the patches that most\nlikely cause the job failure.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/873390\n[2] https://meetings.opendev.org/meetings/releaseteam/2023/releaseteam.2023-03-17-14.00.log.html#l-133\n\nChange-Id: I80ea7da0fe332b7fed36bbf28c0edf280ff560e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/974766a26e29aa91a47edafc6ac1531f94650db2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/974766a26e29aa91a47edafc6ac1531f94650db2"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
