)]}'
{"id":"openstack%2Foslo.log~877817","triplet_id":"openstack%2Foslo.log~stable%2F2023.1~I77377b004f2bb2461b3e61116e85015df7eb74a7","project":"openstack/oslo.log","branch":"stable/2023.1","topic":"revert-oslo-5.1.0-functional-content","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-03-20 16:26:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I77377b004f2bb2461b3e61116e85015df7eb74a7","subject":"Revert \"Fix logging in eventlet native threads\"","status":"MERGED","created":"2023-03-17 15:50:55.000000000","updated":"2023-03-20 16:27:50.000000000","submitted":"2023-03-20 16:26:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877817-revert-oslo-5.1.0-functional-content","meta_rev_id":"c8373647378d8fb56b4b093a8028044897b54be3","_number":877817,"virtual_id_number":877817,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-20 16:26:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-03-20 11:03:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-03-20 09:46:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-03-20 14:49:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-17 17:30:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-20 09:46:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-03-20 11:03:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"ef70ad4f096e5ae8cf4b59f9c1532bdadfbcb0d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-17 15:50:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e49d8d0854fcb76b8e5581ec3831313ed804befb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 17:30:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09ca57927dd149e399299f8cec99b3f0\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5a2e084485de44b0abffde39819107e6 : SUCCESS in 2m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/718aecea1a0e4d8baa4d51c08bb31cbc : SUCCESS in 1h 33m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd76957fd4e7445eb90f807ed32a1967 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/750bbba9f5bf4c4baa1f344799078b44 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b48f5ff1ea7c490d9b9fc80d7a271b85 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1648ed0900a40bb965ef2f1c9c10745 : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b705ef33ab414ed9a9394e92bfd125c3 : SUCCESS in 4m 05s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/f9f2af57719642a2b00ede4039ba3171 : SUCCESS in 1h 07m 21s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/9dad81b8322b46dfb7c7e548c1c0c2a9 : SUCCESS in 30m 46s","accounts_in_message":[],"_revision_number":1},{"id":"dae7b10a2416e4027c5dfb72650bac63f601dfde","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-20 09:46:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c292f1ad1c7ca8b99a02c2bdbebe69a3efbecac","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-03-20 11:03:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ecd0b75f24594753b7ccd42575faf3c82c49e3e1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-20 14:49:53.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db26b2e5754ca701864bf11ee99cd647657b793b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 14:50:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"daa4abdf4448007dde3a0f56a424244d61c02e27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 16:26:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7558cc51aa73488fa698ff058bc4d8dc\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8a47c9bbca084dc88b0dda795d31bf6b : SUCCESS in 1m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/954c2487613c4a77b59d466583788078 : SUCCESS in 1h 29m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/185f0c3397ea42128ac571e26bbfee60 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b1dc7e8b72c440895dd42573e9a400e : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32d62e67c6564f71b9148bde6fac1e69 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a2d5d22358c48a5b22afaff81aaa377 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05571083ce03465d973476addb577e0b : SUCCESS in 2m 52s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/2306b7480ff143e3809cfac71af25127 : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4e4c5c620eea6f4fa0d75ef226d6e9eba98ec5b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 16:26:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c8373647378d8fb56b4b093a8028044897b54be3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 16:27:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/188d9a35f6f940c7bf711cde829fa7fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b4f545736494e3d965094cf13a2b2a8 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7244f15c0f51404ca9f058924dde2032 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9350a409ad7d926cfa17d179f81776097836d0fe","revisions":{"9350a409ad7d926cfa17d179f81776097836d0fe":{"kind":"REWORK","_number":1,"created":"2023-03-17 15:50:55.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/17/877817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/17/877817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/17/877817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/17/877817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/17/877817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/17/877817/1"}}},"commit":{"parents":[{"commit":"974766a26e29aa91a47edafc6ac1531f94650db2","subject":"Revert \"Make the eventlet logging fix execution conditional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/974766a26e29aa91a47edafc6ac1531f94650db2"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-03-17 15:31:23.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-03-17 15:33:33.000000000","tz":60},"subject":"Revert \"Fix logging in eventlet native threads\"","message":"Revert \"Fix logging in eventlet native threads\"\n\nThis reverts commit 94b9dc32ec1f52a582adbd97fe2847f7c87d6c17.\n\nDuring 2023.1 Antelope cycle the upper constraint version bump patch\ncould not merge [1], as cross-nova-py310 fails. This means that during\nthe cycle everything was tested with oslo.log 5.0.0, so as we discussed\nthis during release meeting [2] we need to revert the patches that most\nlikely cause the job failure.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/873390\n[2] https://meetings.opendev.org/meetings/releaseteam/2023/releaseteam.2023-03-17-14.00.log.html#l-133\n\nChange-Id: I77377b004f2bb2461b3e61116e85015df7eb74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/9350a409ad7d926cfa17d179f81776097836d0fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/9350a409ad7d926cfa17d179f81776097836d0fe"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
