)]}'
{"id":"openstack%2Foslo.log~883663","triplet_id":"openstack%2Foslo.log~master~I113d97b3719cce3c4c19c16ea7b8b395cc3beaa0","project":"openstack/oslo.log","branch":"master","topic":"oslo-bump-bandit","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-08-10 17:13:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I113d97b3719cce3c4c19c16ea7b8b395cc3beaa0","subject":"Bump bandit","status":"MERGED","created":"2023-05-19 12:45:44.000000000","updated":"2023-08-10 17:14:20.000000000","submitted":"2023-08-10 17:13:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883663-oslo-bump-bandit","meta_rev_id":"0dcb02309326016f9a3d059f5513b670aa4e7d7f","_number":883663,"virtual_id_number":883663,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-10 17:13:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-06-27 12:06:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-06-27 12:06:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-19 14:17:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-27 12:06:08.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"22d07686cc91010ec0df87a4b2372f86f7d38e11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-19 12:45:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d7f5280061204ea81ae5b1139d6b66d1e600359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 14:17:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d52dcf12318437f8e03a43e663baa36\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a298a75345c1483eacb6d8d92a631761 : SUCCESS in 1m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d84df5c9b264cf5a7f5735f4cc75314 : SUCCESS in 1h 26m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf426452605a48ad89f5826962a1143a : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8c3f41356384c7db44ea9a1e416adb7 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4445e67738e4498a8d25ac992449be60 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b979595109d41b18b14e2fdbaa30b2f : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1f84859a24a433aaa853029afcf7df7 : SUCCESS in 2m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cdc778146354c7b9e58ddf3cbb15d03 : SUCCESS in 2m 31s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/ab899b30af1d49a1aa8398662d877a5b : SUCCESS in 27m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a8866619efd6f157bcd67bc5e9651f789eaffbd1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-27 12:06:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fac980b30ec857d7e3d1e5adf6ad1041fa39c133","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 12:06:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"280f02e36177ca8f7e977ad8a64b1c5d1042051a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 14:30:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2994a62f611f44e7904739341653863a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3737352382c24abca7f8faebca54e346 : SUCCESS in 1m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed620c60e8a14bce8e7084d7c6ad3eac : TIMED_OUT in 2h 04m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/017c0af7223740e791a7d73fd05e147b : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aec7ebf0a23d4cb987a61d193e88c009 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6fa42dcc0d44ddcb32029405855ca9c : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5259c9a72fab46d38fc9157996ce11fa : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afbae445530e44cd9453f3c64c2e8d44 : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37e9631dd56c4fb3ab78e392d60bedc4 : SUCCESS in 3m 21s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/8fdd61d953574d7197e8bb1122d5c50e : SUCCESS in 59m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d37e19564289f4ff9882e54817da0e805ef1f428","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-10 13:37:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"388b2cc24e39218800bfdd4a651aad6662716d56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 15:30:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6455e721e2842f3b303510eef739ce8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6ac0235882054a8e8bfd3eb786c09a32 : SUCCESS in 1m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e2e4761cac148d581905223eb0a0037 : SUCCESS in 1h 45m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4fbdafb6d7143058cd2d456fc2e1dae : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0df7d08dcb194cfb886f1a7695cb058c : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40070039be654b70bc527f9c85a39253 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e91a853525e4948ae576bb28ff6e507 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b20047fcb14a4ed896e074e1cbd26939 : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10dac1f776c045f8b46f9a69d3d9b7ed : SUCCESS in 2m 42s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/582182312dff4e539da219ea7be3cb30 : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f68056c2d28b4ee62dd55bc442c479084d7ee834","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 15:30:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52270126e3fe28a901bb32872832f53fa5470c42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 17:13:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1c347535a2b4e17ac4777b2bc3e2663\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f9f0183c66447c18dfdb33f59bc4013 : SUCCESS in 1m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e81858c550274b3c9e8c576798255a4c : SUCCESS in 1h 37m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/892ecceafe1346e5b54cda0709ccecf7 : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49c47a4b54324518a4f7263a9e5b0d69 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2f000617df142088e420e8d0cf6ca70 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9da90e096c5349a19d831570cc93e9a7 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/996454874c9e40c3bf6b943a94a3fd15 : SUCCESS in 2m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8d839fc103a4c4e896b850403307d85 : SUCCESS in 2m 58s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/8df0cbfc072c4f91809f11c3d4f0867e : SUCCESS in 29m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a7a4a31079dc0b95fe67d42b3fd6bfc3b104bf8e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 17:13:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0dcb02309326016f9a3d059f5513b670aa4e7d7f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 17:14:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e22b714387b4915afb5cefc495161b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e7e7bfdae8490b99e5a7e76d9223f3 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13acee7ae1e3443483e47e712a6e5c4b : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf52b05995658eb7cc6242bb4599b71f23733682","revisions":{"bf52b05995658eb7cc6242bb4599b71f23733682":{"kind":"REWORK","_number":1,"created":"2023-05-19 12:45:44.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/63/883663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/63/883663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/63/883663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/63/883663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/63/883663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/63/883663/1"}}},"commit":{"parents":[{"commit":"a0c3ca8ce1f6c58069be4dfbe2bd19c33b526c6c","subject":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a0c3ca8ce1f6c58069be4dfbe2bd19c33b526c6c"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-05-19 12:45:39.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-05-19 12:45:39.000000000","tz":120},"subject":"Bump bandit","message":"Bump bandit\n\nChange-Id: I113d97b3719cce3c4c19c16ea7b8b395cc3beaa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/bf52b05995658eb7cc6242bb4599b71f23733682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/bf52b05995658eb7cc6242bb4599b71f23733682"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
