)]}'
{"id":"openstack%2Foslo.log~890917","triplet_id":"openstack%2Foslo.log~master~I7631394bee7281f6a393ab4ba7604f871626997a","project":"openstack/oslo.log","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-08-10 13:36:49.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2023-08-23 08:45:02.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-08-23 08:45:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7631394bee7281f6a393ab4ba7604f871626997a","subject":"Catch RuntimeError when loading log config file","status":"MERGED","created":"2023-08-09 14:03:33.000000000","updated":"2023-08-23 08:46:06.000000000","submitted":"2023-08-23 08:45:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890917","meta_rev_id":"4b887c3c1925f694aaa928e83116e0ca7e92b528","_number":890917,"virtual_id_number":890917,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-23 08:45:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-08-10 13:36:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-08-23 07:06:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-08-23 07:06:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-09 15:16:45.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-08-09 15:48:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-10 13:06:18.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-08-23 07:06:41.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"dd08b2d17971e8986e98e26d1e91e74555eff5ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-08-09 14:03:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a3298493f9951db839b209970207659ef372aa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 15:48:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/118b60a0142a406b89cffd50f63d5a38\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74842704f71d4f9f8330a3796503b861 : SUCCESS in 1h 37m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cd4cb06c1cc40938a111a9358d9580d : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd3ff1bb31e0461098c37c7652750a47 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88182ca9b9ac43869477722101cc64b3 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1408fbd34a242fa860e2651b0f34ab4 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41d3e7cbabd24729980b2a1e6a1f6827 : SUCCESS in 3m 30s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/4983dbbb5eb94eaf8ab02d80f018a9eb : SUCCESS in 1h 22m 14s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/180b26e8713a42149ab44aefc0ca8d15 : SUCCESS in 40m 17s","accounts_in_message":[],"_revision_number":1},{"id":"fc0b709b41559bc448626d2ed873f243e7d52bbc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-10 13:36:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"70d3857ac8600e69a1e2347ab60546f1f419f014","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-23 07:06:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f214648c241a29fe07fb2d9b1fc2986683c60bc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 07:07:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd60dfc2ee2dc0cbc4b3ea3b30ebd9651c873823","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 08:45:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f274c2984226491ea4a0c36dce7470e9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88e49a4df1324672b4d973da7b819090 : SUCCESS in 1h 32m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2536fc15efd486ca72f72c2d274f694 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92685b3cc3be499c8ac0c39fa3d44786 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/030bca47359b4848a9e5338de4c96f99 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b2179807b144134a080aa31b6ad53a0 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66d6de9b28ca4703ab5443cce6d5c6a2 : SUCCESS in 2m 55s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/10ae18d256194e409099e7b6cfd3e4b7 : SUCCESS in 25m 49s","accounts_in_message":[],"_revision_number":1},{"id":"82e2ce06d964ed39702c3751ad1d5f0801128a51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 08:45:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4b887c3c1925f694aaa928e83116e0ca7e92b528","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 08:46:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9752d352800476bb2467acf6b1a40de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c1a468abfb441589a81f976c3c1e7d : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"390a1d108eb8717f8471098b92947cb7dfd2cdf0","revisions":{"390a1d108eb8717f8471098b92947cb7dfd2cdf0":{"kind":"REWORK","_number":1,"created":"2023-08-09 14:03:33.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/17/890917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/17/890917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/17/890917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/17/890917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/17/890917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/17/890917/1"}}},"commit":{"parents":[{"commit":"eaf6089ab23d347dbe5ca660db74d646064269ae","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/eaf6089ab23d347dbe5ca660db74d646064269ae"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-08-09 10:40:17.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-08-09 14:03:23.000000000","tz":120},"subject":"Catch RuntimeError when loading log config file","message":"Catch RuntimeError when loading log config file\n\nSince python 3.12, logging is raising RuntimeError when content is\ninvalid [1][2]. Note this is also backported to 3.11 branch.\n\nThis is making oslo_log to not raise LogConfigError but throw the\nunderlying RuntimeError.\n\n[1] https://github.com/python/cpython/commit/152227b569c3a9b87fe0483706f704762ced6d75\n[2] https://github.com/python/cpython/commit/12f1581b0c43f40a792c188e17d2dea2357debb3\n\nChange-Id: I7631394bee7281f6a393ab4ba7604f871626997a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/390a1d108eb8717f8471098b92947cb7dfd2cdf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/390a1d108eb8717f8471098b92947cb7dfd2cdf0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
