)]}'
{"id":"openstack%2Foslo.log~896919","triplet_id":"openstack%2Foslo.log~master~I2a9a2ab4b03e7d6b804a375c877c2c632df11945","project":"openstack/oslo.log","branch":"master","topic":"caracal_versionutils","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2026-03-08 14:43:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2a9a2ab4b03e7d6b804a375c877c2c632df11945","subject":"Add Antelope, Bobcat and Caracal releases to versionutils","status":"ABANDONED","created":"2023-09-29 09:29:07.000000000","updated":"2026-03-08 14:43:41.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2c21559e27631a8e26107b7e53764371524348d5","_number":896919,"virtual_id_number":896919,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-29 11:07:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-2,"date":"2026-01-05 14:46:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-29 11:07:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-05 14:46:33.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"2402613eb6bc6c0e0cec73dfbc6d4fbf92ea74b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-29 09:29:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7165848ff2d2225a4059596eb4cd6a2f1ae7cbd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:07:14.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/358e2f26c5d6492dad8ec362298ad9c4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d369ea3db859450eabb42f59a1c55c27 : SUCCESS in 1h 37m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c31506529d294524815061646f39535d : FAILURE in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af9cc0606bad4ba889bd150c630690da : FAILURE in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c779d19dc5504fe1b6ae256eadeaadb1 : FAILURE in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27837b837cb548ac8cc4acc321b72767 : FAILURE in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0b2454da9254e2790ad6d058667d50c : FAILURE in 3m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/721ee01cc64c44f9a5e7ff972a6c0e28 : SUCCESS in 3m 14s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/028ec8c955354f579047e5c4ee608550 : SUCCESS in 58m 49s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/05d0a8548a894417a174c54fcfcdea9a : SUCCESS in 38m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9688c7d843de2a06c80384ad86ceb1c949ecb6fc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-05 14:46:33.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c21559e27631a8e26107b7e53764371524348d5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-08 14:43:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d5630cd67e4b6f279cef01c255e863b4f3de208","revisions":{"9d5630cd67e4b6f279cef01c255e863b4f3de208":{"kind":"REWORK","_number":1,"created":"2023-09-29 09:29:07.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/19/896919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/19/896919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/19/896919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/19/896919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/19/896919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/19/896919/1"}}},"commit":{"parents":[{"commit":"b5b8c30b0d925aa3d31b58932c94586631827b62","subject":"Merge \"Deprecate Windows support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b5b8c30b0d925aa3d31b58932c94586631827b62"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-09-29 09:26:56.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-09-29 09:26:56.000000000","tz":120},"subject":"Add Antelope, Bobcat and Caracal releases to versionutils","message":"Add Antelope, Bobcat and Caracal releases to versionutils\n\nNow that we have names for the next release we can get it\nadded here so we\u0027ll be ready when development starts.\n\nAlso adding previous series names. They were missing.\n\nChange-Id: I2a9a2ab4b03e7d6b804a375c877c2c632df11945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/9d5630cd67e4b6f279cef01c255e863b4f3de208"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/9d5630cd67e4b6f279cef01c255e863b4f3de208"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
