)]}'
{"id":"openstack%2Foslo.log~905963","triplet_id":"openstack%2Foslo.log~master~I1d9544123b05f2c486eef4e6ae6ed9e6e468ca93","project":"openstack/oslo.log","branch":"master","topic":"bump-eventlet","hashtags":[],"change_id":"I1d9544123b05f2c486eef4e6ae6ed9e6e468ca93","subject":"bump eventlet to latest version that support python 3.12","status":"ABANDONED","created":"2024-01-17 15:48:40.000000000","updated":"2024-01-19 12:51:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b836f7edc01337bdb57e191fc13b773794e07740","_number":905963,"virtual_id_number":905963,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-17 18:04:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-17 18:04:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88e236436037403269610c71c91a95b3f0b2567c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-17 15:48:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"268775c3641fd49f5f2d04398c185ee8ced03e7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:04:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a428c740c66412aacff1eb8749dd93e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9e95055281064aa69c5b68d46b78afc6 : SUCCESS in 2m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e597e816372e4720969e6198fea22e20 : SUCCESS in 1h 35m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1f336bfa0014e859fa2477e3e9e7ca7 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cac187abce5e4c72b801c3439154ffa4 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a5019ed0f4b48f5bcd08c8b1fe8c32b : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8796d9436c243dabc6842c83bc4be27 : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/130fe9235dd94ef0b8ef778eb63be788 : SUCCESS in 3m 59s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/aa88cd10af81496ca301376ee9aea322 : SUCCESS in 33m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b836f7edc01337bdb57e191fc13b773794e07740","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-19 12:51:06.000000000","message":"Abandoned\n\nThis version have been reverted from Upper Constraints. https://review.opendev.org/c/openstack/requirements/+/905905","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f9ce6e7dd9885d521a16de7614f5e10bf82985e","revisions":{"3f9ce6e7dd9885d521a16de7614f5e10bf82985e":{"kind":"REWORK","_number":1,"created":"2024-01-17 15:48:40.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/63/905963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/63/905963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/63/905963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/63/905963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/63/905963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/63/905963/1"}}},"commit":{"parents":[{"commit":"fddcbf13651a355ed020e2ad7093037312788287","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/fddcbf13651a355ed020e2ad7093037312788287"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-01-17 15:48:26.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-01-17 15:48:37.000000000","tz":60},"subject":"bump eventlet to latest version that support python 3.12","message":"bump eventlet to latest version that support python 3.12\n\nChange-Id: I1d9544123b05f2c486eef4e6ae6ed9e6e468ca93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3f9ce6e7dd9885d521a16de7614f5e10bf82985e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3f9ce6e7dd9885d521a16de7614f5e10bf82985e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
