)]}'
{"id":"openstack%2Foslo.log~906609","triplet_id":"openstack%2Foslo.log~master~If6b2d64febbf00814503eccd59dd9dadd330e0ba","project":"openstack/oslo.log","branch":"master","topic":"bump-hacking","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-06 08:27:53.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"If6b2d64febbf00814503eccd59dd9dadd330e0ba","subject":"Bump hacking","status":"MERGED","created":"2024-01-25 03:18:31.000000000","updated":"2024-02-06 17:49:33.000000000","submitted":"2024-02-06 17:43:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906609-bump-hacking","meta_rev_id":"e254787dda8d10c9fd8b92eb1f95578fc367e3ac","_number":906609,"virtual_id_number":906609,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-06 17:43:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-02-06 15:55:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-02-06 15:55:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-25 05:02:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e7e259678d1a9ddd1cf9543dcfd204903621dc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-25 03:18:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b6be689a69b40978e1b721ae42c9d24504eec62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 05:02:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec2d0103686348519b0608a3b26a73ef\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/20887dd4509b464eadb88532569bfa22 : SUCCESS in 1m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/284a68e9ac68477dad9b10454d5eaec9 : SUCCESS in 1h 26m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bc682d6fb5246778d0c239585787410 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85c87875d85245009e3035fce550afc4 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33018dfcf20f4ba794adc5c95dba9817 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38568ef7808848d1939dfd9d2b7c1aa3 : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb012b9849d1449fb80fd9d99de7b7ff : SUCCESS in 2m 59s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/450b547f45f74e0bac7f324637a5c1dc : SUCCESS in 36m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5e53845f9bca314bdf4fb45618c4ae48f534de76","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 06:16:05.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4d5dbde2893076cdaab508b8801d2017606054eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 06:16:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e94e3f030145023cb659ecd555447f5e3820d84b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 07:39:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff90f75e101d47e495f62a7a7ac94dc2\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/64e2ae04b50d42e1bef8b42f360484a1 : SUCCESS in 1m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f73ff1029e0f493bb12c601de4acb63b : FAILURE in 1h 15m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b83930c4cf843f4b3f175f2894c0f76 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e812837fcbc343c58aedffcddcfb5567 : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73814048e06840e1938c48dd5695813d : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71a2cf0c48c04e6889a433164eb6f37e : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3746d868b04d4ebf948ef1be36d9b003 : SUCCESS in 2m 41s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/6df4af4e08694fab99cde23b58562370 : SUCCESS in 1h 01m 36s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/05ac4ad02b1e48b8a9bde5c7f39a7529 : SUCCESS in 37m 09s","accounts_in_message":[],"_revision_number":2},{"id":"8f6564b8ca79711250355665e2afd81b5fb6acac","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 08:27:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"535bf4bcd6de1d72c8ffd9227221aefafdcc47e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 10:12:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51b43997a06c46e69a748259174787a4\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6aeecff7c3344802986c63885ef962bf : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a7f3e2c278e486fb2620fa29892b532 : SUCCESS in 1h 34m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44349979937e4f879316cdfe29ee7e8c : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3be62921e10c4b01ae4b4b7f50d5aa1b : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b64efad97614576a2a696d12a404a26 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bee794ce465644b4afb16e43459ed988 : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e47a2fdc0d2a4634b216844e23edcf20 : SUCCESS in 2m 44s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/165113bbc6e24d568ba434d2bebd060e : SUCCESS in 1h 09m 18s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/b8ffc9e3c8f34a0fb70e32371212f89c : SUCCESS in 35m 07s","accounts_in_message":[],"_revision_number":2},{"id":"7430cef9f08ff5ca2409bd0d1e2a19307c161fa8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 15:55:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"18089c11c451528228c7495fecbb54252feb455a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 15:56:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fd22a17716595da67b74b9973736121b15f3c6d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 17:43:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88985c6a822c4ff5a312b3d86383ac48\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d7f618abef0444789953b9e3714959d4 : SUCCESS in 2m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e24d0c223944715bcedc9818e7e603e : SUCCESS in 1h 40m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70e3c314fd534cda8f68b15434e585ab : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f89e2031e31473f858b9ce8a0449113 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc3cd80635f54a75b41f92eb66eb3745 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/054151ddb2a446d69883f29cc8fce923 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb7f0d05ddf8489783456317f82948ce : SUCCESS in 3m 04s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/c029719de3974c88b8b2970f5a1b9e1e : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":2},{"id":"70107ac57b7c017a97fe1f53fe4d8b89a1f29ac4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 17:43:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e254787dda8d10c9fd8b92eb1f95578fc367e3ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 17:49:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f402aa3a987b46138bafea85a29323e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82cfb997249a4cf2b3ec271cb1c056df : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3237a49ad0a41b19950f40c5960b5b2 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"631b2f62dcc3152a7a1bd08e3db1330c13e372f5","revisions":{"2da752a98ae12e5d0f67ffefb44e808bbb2d248a":{"kind":"REWORK","_number":1,"created":"2024-01-25 03:18:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/906609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/09/906609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/09/906609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/09/906609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/09/906609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/09/906609/1"}}},"commit":{"parents":[{"commit":"fddcbf13651a355ed020e2ad7093037312788287","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/fddcbf13651a355ed020e2ad7093037312788287"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-25 03:17:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-25 03:18:21.000000000","tz":540},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 2.0.x is too old.\n\nChange-Id: If6b2d64febbf00814503eccd59dd9dadd330e0ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2da752a98ae12e5d0f67ffefb44e808bbb2d248a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2da752a98ae12e5d0f67ffefb44e808bbb2d248a"}]},"branch":"refs/heads/master"},"631b2f62dcc3152a7a1bd08e3db1330c13e372f5":{"kind":"REWORK","_number":2,"created":"2024-02-06 06:16:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/906609/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/09/906609/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/09/906609/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/09/906609/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/09/906609/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/09/906609/2"}}},"commit":{"parents":[{"commit":"fddcbf13651a355ed020e2ad7093037312788287","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/fddcbf13651a355ed020e2ad7093037312788287"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-25 03:17:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 06:16:17.000000000","tz":540},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 2.0.x or 3.0.x is too old.\n\nChange-Id: If6b2d64febbf00814503eccd59dd9dadd330e0ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/631b2f62dcc3152a7a1bd08e3db1330c13e372f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/631b2f62dcc3152a7a1bd08e3db1330c13e372f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
