)]}'
{"id":"openstack%2Foslo.log~907947","triplet_id":"openstack%2Foslo.log~master~Ib7af2d4299853e9758cbf148908aa1ef5de1b235","project":"openstack/oslo.log","branch":"master","topic":"pre-commit","hashtags":[],"change_id":"Ib7af2d4299853e9758cbf148908aa1ef5de1b235","subject":"pre-commit: Bump versions","status":"MERGED","created":"2024-02-06 06:19:09.000000000","updated":"2024-02-21 05:49:56.000000000","submitted":"2024-02-21 05:48:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907947-pre-commit","meta_rev_id":"1cc770e1c55cdc0ace22c6dd5588d37d22f5ec6d","_number":907947,"virtual_id_number":907947,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-21 05:48:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-02-21 04:18:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-02-21 04:18:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-06 08:08:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f14d1885fe4b4ff22658cf4dd8403e710769468d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 06:19:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fbd79c2808ef4a56e4712b5260cd5ef1bcac389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 08:08:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/228e2df229db4d5fb0227471a30aa1f3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a47de2894174291abab23e004424b56 : SUCCESS in 1m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77557cd1d2fd4caa9c1897044772426a : SUCCESS in 1h 40m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4b02651f79c49deb615fcc8d97cf5ac : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49038e7ac19d452d8d0d2b627f3fd9f6 : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/275298d078d047b0a3132110523e6db4 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ec004b80a274b31acf5ca89499e187d : SUCCESS in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f39939b91ffa4694bd016c25bd3b939b : SUCCESS in 4m 27s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/60c5f4afad6a46d99ea4b067946bba1c : SUCCESS in 47m 13s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/eeafba2efdd843499a4e7eae5cf3c145 : SUCCESS in 29m 53s","accounts_in_message":[],"_revision_number":1},{"id":"594ccf238957e07ba011fdfb834727e418579640","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 04:18:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c29d93391de36aa10fb3c889e9d3e4763cf3129","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 04:19:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4953c6bb8a9a23a5ed33fec18eb4c819463e95fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 05:48:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2875ee5c83874190b6f799e099dbe275\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f0a504a63925439694299b0eb0db6275 : SUCCESS in 2m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71dc7d9476b746efaa14cd7da21f6a90 : SUCCESS in 1h 29m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b491c8b4603a4c35b31d769920f75aaa : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04060842f97a4613aa5b3b48b3e98f60 : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be69045fe18643e6ab87d2479e873306 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b581fa23856e4aacaeb898964c017ab3 : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00aef2943bb64fad92424a4fd624619e : SUCCESS in 3m 04s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/866385d5f48c41d385660a84f0e0487d : SUCCESS in 24m 37s","accounts_in_message":[],"_revision_number":1},{"id":"062d1f4d89e148f47fce48bb425702c6ec287833","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 05:48:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1cc770e1c55cdc0ace22c6dd5588d37d22f5ec6d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 05:49:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d33000e04914a15b1253a855e95aec0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5b716f4b32e4b22b8a5290d0aba8b7b : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d9bc7cbfd34459e90aac9957a6bf401 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de07539a7b93f307bc6ee2fe54734f852ac746ec","revisions":{"de07539a7b93f307bc6ee2fe54734f852ac746ec":{"kind":"REWORK","_number":1,"created":"2024-02-06 06:19:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/907947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/47/907947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/47/907947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/47/907947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/47/907947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/47/907947/1"}}},"commit":{"parents":[{"commit":"631b2f62dcc3152a7a1bd08e3db1330c13e372f5","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/631b2f62dcc3152a7a1bd08e3db1330c13e372f5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 06:17:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 06:17:51.000000000","tz":540},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nhacking now has a native hook.\n\nCo-Authored-By: Stephen Finucane \u003csfinucan@redhat.com\u003e\nChange-Id: Ib7af2d4299853e9758cbf148908aa1ef5de1b235\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/de07539a7b93f307bc6ee2fe54734f852ac746ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/de07539a7b93f307bc6ee2fe54734f852ac746ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
