)]}'
{"id":"openstack%2Foslo.log~914190","triplet_id":"openstack%2Foslo.log~master~I860abe953ce945bb5152c77a7daeb6aa1003512b","project":"openstack/oslo.log","branch":"master","topic":"bug/2039346","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-26 18:27:08.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-03-26 14:55:29.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2024-03-26 17:17:25.000000000","reason":"\u003cGERRIT_ACCOUNT_14525\u003e replied on the change","reason_account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}}},"hashtags":[],"change_id":"I860abe953ce945bb5152c77a7daeb6aa1003512b","subject":"Fix eventlet detection","status":"MERGED","created":"2024-03-26 10:38:25.000000000","updated":"2024-03-26 18:28:13.000000000","submitted":"2024-03-26 18:27:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914190-bug/2039346","meta_rev_id":"de169b856f2ee3a8d3ea76d61859adbf461bf6b7","_number":914190,"virtual_id_number":914190,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-26 18:27:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-03-26 14:51:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-03-26 14:55:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-03-26 14:51:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-03-26 14:51:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-03-26 14:55:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-26 11:06:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2024-03-26 13:17:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 13:59:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-03-26 14:43:20.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-03-26 15:19:21.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"state":"CC"},{"updated":"2024-03-26 15:19:21.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"}],"messages":[{"id":"a1fb8a0db7798f13f12f0738da87213e766555ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-03-26 10:38:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eeb517808bd2dbb8357f841e3a8b434543a3fb1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-03-26 10:49:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"669117e65402c03e5f0c469cbe61028a422bc5a1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-03-26 11:06:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4c294243ff9f23dbc1733897e8c700571e8ec150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-03-26 12:02:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ab624ac019b679d7aae5dff18c025b32813b0037","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-03-26 12:03:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cb9a8430e013b718ad6075ab9d0d4e90410c3f46","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-03-26 12:03:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"883ac454f495fd8658b0c571631d7a5a84d3cb0d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-03-26 13:06:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"99486cd37fe6aa3f126400d5bc8c50418f395ec4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-03-26 13:07:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"42174164a17116b20ea21edf85a28b57b4746e90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 13:17:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/494fa18701c5401e86be8b008a144120\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/296d927f27c14a66b32d2f12f1c064d1 : SUCCESS in 1h 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ab8a5d85d94a73be4122a5a2a497b9 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1c20504364e479c8d731899652bc6f9 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c4e9543ca7e44a97bc0739236236299a : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0fd3bc88813847eb9b67e967418d05f3 : FAILURE in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dff5d7d59f6491ea526ddb58ffc70c2 : SUCCESS in 4m 18s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/f63dd48805c44f0988e0b81e5d197312 : SUCCESS in 1h 02m 26s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/146d5416216644e0b1022c1001327bf8 : SUCCESS in 41m 26s","accounts_in_message":[],"_revision_number":3},{"id":"c94d2626f03a2a039ae04ada0e12b68523d8f5f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-03-26 13:47:47.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"54cc8b14bcd33308db9f3872e850b865b2d0e326","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-03-26 13:51:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c69fcebe32357a7f057a3fb04d087656096237a5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-03-26 13:59:25.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a741e18b291704f6c29cf9974e88bf51c3675baf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-03-26 14:00:28.000000000","message":"Patch Set 4: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d2ec5d3a0ea2511a6a30b3faaaf7011615357447","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-26 14:40:16.000000000","message":"Topic mal removed","accounts_in_message":[],"_revision_number":4},{"id":"98a4d4e07c80fb207d0c67451c4beb3fb2e40190","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-26 14:41:06.000000000","message":"Topic set to bug/2039346","accounts_in_message":[],"_revision_number":4},{"id":"cd82c4fa4debea744bdb9b38fb2c333f7f5af530","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-26 14:43:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3f99aac0fe84635c6e49c0a38ca56ecf8a5c39d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-26 14:50:12.000000000","message":"Patch Set 5: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"1733631f0dcd6dc983e105909368b3ad6ec24a75","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-26 14:51:03.000000000","message":"Patch Set 5: Backport-Candidate+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"44b1ff299d5b2f3271763a8c482cd3fc325e2c78","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-03-26 14:55:29.000000000","message":"Patch Set 5: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e55402affafc4999a0edee0e76b3755f729a8534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 16:29:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fc2660cd40542f8b373f03c325ba837\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/965b20aff44c448e9a6241c38491a376 : SUCCESS in 1h 32m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7d7a6e3a34442438b86bdab942f3290 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e606b781c134cfead442d0b80ec3d82 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31cde35c32e94f3fae2955ae88fddc22 : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b3d471e2bf94f76b189ff853de33fe1 : FAILURE in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e265b6afdb3b42e0898087fa0d181cbf : SUCCESS in 3m 28s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/04ac3a9a862540028bf33b6de259366d : SUCCESS in 1h 04m 15s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/72ba9977160949298baf2dc53e0215c5 : SUCCESS in 28m 33s","accounts_in_message":[],"_revision_number":5},{"id":"44d6e09c5eba655fd16db1c8066c5f8403ac4a56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 16:30:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"453749612a64b98b37756056391ac5711d568659","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-03-26 17:17:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"45cbd6f109aae617ddf1e2a0321520116e5d6aa3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 18:27:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a094e9923034ff190be1a3a0f7c62c1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35139ce498ac4f4f9f56d59b2837624d : SUCCESS in 1h 48m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8324e11782541f99bf084a3b2c6ec5e : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5994f9eddba14cac923c1aa82a38a579 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/edbee13d5b754160bb92ea8227f58e19 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a1bbcdeec24a10b65b5b1e595daf93 : SUCCESS in 3m 16s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/dc90ecc85f5945e4bc266c620e5f1936 : SUCCESS in 29m 24s","accounts_in_message":[],"_revision_number":5},{"id":"cc33eb2fb1e5da7bd69573efd8d105de68057148","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 18:27:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"de169b856f2ee3a8d3ea76d61859adbf461bf6b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 18:28:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/096c8e1340ea47ba82806b7e6997f5a0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de7ebc0d476943c4a332dffaeeb46cdb : SUCCESS in 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a1fe1b9cfb841d632aa582a52c106b4ebab1c159","revisions":{"605cc9b6244ae61c8f4b84842de6d14d0a5aaf6a":{"kind":"REWORK","_number":1,"created":"2024-03-26 10:38:25.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/90/914190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/90/914190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/1"}}},"commit":{"parents":[{"commit":"3e58709e8035d32bb37415b665b8a49dc89552e2","subject":"Merge \"reno: Update master for unmaintained/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3e58709e8035d32bb37415b665b8a49dc89552e2"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-03-26 10:33:16.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-03-26 10:37:17.000000000","tz":120},"subject":"Fix eventlet detection","message":"Fix eventlet detection\n\nEventlet may be installed, but not used for example projects\nlike octavia. Improve autodetection mechanism by trying to import\nmodule and check if it is actually patched.\n\nChange-Id: I860abe953ce945bb5152c77a7daeb6aa1003512b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/605cc9b6244ae61c8f4b84842de6d14d0a5aaf6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/605cc9b6244ae61c8f4b84842de6d14d0a5aaf6a"}]},"branch":"refs/heads/master"},"a74b28364e7f6cbcad6e3af739d73a0588cdea6f":{"kind":"REWORK","_number":2,"created":"2024-03-26 10:49:14.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/90/914190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/90/914190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/2"}}},"commit":{"parents":[{"commit":"3e58709e8035d32bb37415b665b8a49dc89552e2","subject":"Merge \"reno: Update master for unmaintained/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3e58709e8035d32bb37415b665b8a49dc89552e2"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-03-26 10:33:16.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-03-26 10:49:07.000000000","tz":120},"subject":"Fix eventlet detection","message":"Fix eventlet detection\n\nEventlet may be installed, but not used for example projects\nlike octavia. Improve autodetection mechanism by trying to import\nmodule and check if it is actually patched.\n\nChange-Id: I860abe953ce945bb5152c77a7daeb6aa1003512b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a74b28364e7f6cbcad6e3af739d73a0588cdea6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a74b28364e7f6cbcad6e3af739d73a0588cdea6f"}]},"branch":"refs/heads/master"},"0fbaed9d39c93e59bab26f952896088f46bb1793":{"kind":"REWORK","_number":3,"created":"2024-03-26 12:02:24.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/90/914190/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/90/914190/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/3"}}},"commit":{"parents":[{"commit":"3e58709e8035d32bb37415b665b8a49dc89552e2","subject":"Merge \"reno: Update master for unmaintained/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3e58709e8035d32bb37415b665b8a49dc89552e2"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-03-26 10:33:16.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-03-26 12:02:08.000000000","tz":120},"subject":"Fix eventlet detection","message":"Fix eventlet detection\n\nEventlet may be installed, but not used for example projects\nlike octavia. Improve autodetection mechanism by trying to import\nmodule and check if it is actually patched.\n\nChange-Id: I860abe953ce945bb5152c77a7daeb6aa1003512b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0fbaed9d39c93e59bab26f952896088f46bb1793"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0fbaed9d39c93e59bab26f952896088f46bb1793"}]},"branch":"refs/heads/master"},"38f8ad79705b5028c859802cfe72e31ef5bc59df":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-03-26 13:47:47.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/90/914190/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/90/914190/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/4"}}},"commit":{"parents":[{"commit":"3e58709e8035d32bb37415b665b8a49dc89552e2","subject":"Merge \"reno: Update master for unmaintained/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3e58709e8035d32bb37415b665b8a49dc89552e2"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-03-26 10:33:16.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-03-26 13:47:47.000000000","tz":0},"subject":"Fix eventlet detection","message":"Fix eventlet detection\n\nEventlet may be installed, but not used for example projects\nlike octavia. Improve autodetection mechanism by trying to import\nmodule and check if it is actually patched.\n\nRelated-Bug: #2039346\n\nChange-Id: I860abe953ce945bb5152c77a7daeb6aa1003512b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/38f8ad79705b5028c859802cfe72e31ef5bc59df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/38f8ad79705b5028c859802cfe72e31ef5bc59df"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a1fe1b9cfb841d632aa582a52c106b4ebab1c159":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-03-26 14:50:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/914190/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/90/914190/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/90/914190/5"}}},"commit":{"parents":[{"commit":"3e58709e8035d32bb37415b665b8a49dc89552e2","subject":"Merge \"reno: Update master for unmaintained/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3e58709e8035d32bb37415b665b8a49dc89552e2"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-03-26 10:33:16.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-26 14:50:12.000000000","tz":0},"subject":"Fix eventlet detection","message":"Fix eventlet detection\n\nEventlet may be installed, but not used for example projects\nlike octavia. Improve autodetection mechanism by trying to import\nmodule and check if it is actually patched.\n\nCloses-Bug: #2039346\nChange-Id: I860abe953ce945bb5152c77a7daeb6aa1003512b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a1fe1b9cfb841d632aa582a52c106b4ebab1c159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a1fe1b9cfb841d632aa582a52c106b4ebab1c159"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
