)]}'
{"id":"openstack%2Foslo.log~914788","triplet_id":"openstack%2Foslo.log~master~If776fee0072d95737e1c19ad673a3c881a89de06","project":"openstack/oslo.log","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-04-02 10:31:05.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-03 14:59:50.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-04-03 14:59:50.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-04-03 13:10:14.000000000","reason":"\u003cGERRIT_ACCOUNT_31245\u003e replied on the change","reason_account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}}},"hashtags":[],"change_id":"If776fee0072d95737e1c19ad673a3c881a89de06","subject":"Deprecate watch_log_file","status":"MERGED","created":"2024-03-30 16:03:19.000000000","updated":"2024-04-03 15:28:30.000000000","submitted":"2024-04-03 14:59:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914788","meta_rev_id":"26f7e6935ae880fe6a2b654e541afa594cbfe136","_number":914788,"virtual_id_number":914788,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-03 14:59:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2024-04-02 10:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-03 13:10:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-03 13:10:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"CC":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}],"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-30 17:27:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-31 05:24:40.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2024-03-31 05:24:40.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-03-31 05:24:40.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2024-04-02 07:37:39.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"}],"messages":[{"id":"b3ec3dc37dc5d7ddd1e2ee7f693590bd64e3658c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-30 16:03:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2e981a99818e062f0a8a827bddcbbbfd2b0eca4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-30 16:05:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4d1482de92b427f8b2ed0affe120e4882f75cf87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-30 16:08:48.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"06b467f8841e48ac35855e8b09f723b88ce66045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 17:27:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/599ce1664c024cedb88ae52eb8871965\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25a00cbdfd764d6b8c94f03b524195e3 : SUCCESS in 58m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aa962843c3b4a43affc6e3036ff30d7 : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00ce43841814466995952898b7e8d357 : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fad566a8bd4040718c41e7b2299fe630 : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38319be19e884a97b45e390f1cc5b830 : FAILURE in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd2c4b4fe5fd4c40806a00b220e80d41 : SUCCESS in 2m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73d441e306d74f57bf3d10151e443b80 : SUCCESS in 2m 23s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/5323b08ecc994005a65f527ee362baf3 : SUCCESS in 1h 08m 13s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/2fb633c031b747e5816dcd37603055aa : SUCCESS in 43m 40s","accounts_in_message":[],"_revision_number":3},{"id":"29ca394fd1f9ffa47a2011386e85c2ff961b02d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-31 05:10:58.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f5a0e9289fd3a49918a3f447a81fe0e1fd3b24cd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-31 05:24:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c7604065db9fc3747620be20a67f38aa4621c4a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-31 07:41:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09ef052e694146b996d53e5f05a61c4d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebe68a41ba0547abb4beadfdcc18aa8b : SUCCESS in 1h 39m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b68c113bb0fc4f348ef59573acadcf73 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/705c253f21d84cfb8f6a747fb8880e3c : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c7da83995294e1988ee3e8ce2f0383a : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7304e785033643cfbf702768e6d05de9 : FAILURE in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7309fc8b8cbd4fd88ddb3b3f84cb4ce4 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0974c6f0ae84a6488fbf8d7d3de461f : SUCCESS in 2m 45s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/eed7f072d768431c8b04cbf24a70b406 : SUCCESS in 51m 24s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/a188645a33664a73ae075b81f600992f : SUCCESS in 38m 05s","accounts_in_message":[],"_revision_number":4},{"id":"95f86a0b19d05fb0939a829c9be76cab3607589b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-02 10:31:05.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"cf06b45421293efd47743e0666b77c99212acef8","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-04-03 13:10:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a6e2438c3b8584ffd3ab92cd723a730a1581a88b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 13:20:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c2b0dba2671a05fde9bd0bf18c2abb5c20128aea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 14:59:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c575e9e335f24dbea55a42518e91de08\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0802b91b79041198af50fc5aa6e15c9 : SUCCESS in 1h 31m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e73bb6ff09ed4efda50b835ab6f72bc5 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d824b17385804898b0999796693ff324 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8a2b167c42948d39c8d95b9d0a1251b : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d7d2b147764cde99789f633974ffba : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c55816d76e8b44a0bbc1e321c85ad8f6 : SUCCESS in 2m 49s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/ff69b681d979477c84a405e4ac80d3f5 : SUCCESS in 43m 15s","accounts_in_message":[],"_revision_number":4},{"id":"5fc97098c015b4266f852dd7d46a703b4a655e5a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 14:59:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"26f7e6935ae880fe6a2b654e541afa594cbfe136","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 15:28:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e38f9dfcde24687b79b4aa7ed0e90f6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b0565bfc9b2451b892dbedb81f18285 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c54850a77c944a787aed0aecf34f032 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"31603ac6dacd0b47ec9f02d99bfdd0a7b683119c","revisions":{"fbc7bf22bce33443168e940628f577041dda33b1":{"kind":"REWORK","_number":1,"created":"2024-03-30 16:03:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/914788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/88/914788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/1"}}},"commit":{"parents":[{"commit":"a1fe1b9cfb841d632aa582a52c106b4ebab1c159","subject":"Fix eventlet detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a1fe1b9cfb841d632aa582a52c106b4ebab1c159"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 15:54:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 15:54:59.000000000","tz":540},"subject":"Deprecate watch_log_file","message":"Deprecate watch_log_file\n\nThere is a known issue reported some time ago[1] but it hasn\u0027t been\nfixed.\n\nThe pyinotify package used hasn\u0027t been updated for 9 years and looks\nbadly unmaintained.\n\nRelated-Bug: #1740111\nChange-Id: If776fee0072d95737e1c19ad673a3c881a89de06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/fbc7bf22bce33443168e940628f577041dda33b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/fbc7bf22bce33443168e940628f577041dda33b1"}]},"branch":"refs/heads/master"},"992a76cf9e97e65fcbf62e72cbd57c7915f04cd9":{"kind":"REWORK","_number":2,"created":"2024-03-30 16:05:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/914788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/88/914788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/2"}}},"commit":{"parents":[{"commit":"a1fe1b9cfb841d632aa582a52c106b4ebab1c159","subject":"Fix eventlet detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a1fe1b9cfb841d632aa582a52c106b4ebab1c159"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 15:54:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 16:04:48.000000000","tz":540},"subject":"Deprecate watch_log_file","message":"Deprecate watch_log_file\n\nThere is a known issue reported some time ago[1] but it hasn\u0027t been\nfixed.\n\nThe pyinotify package used hasn\u0027t been updated for 9 years and looks\nbadly unmaintained.\n\nRelated-Bug: #1740111\nChange-Id: If776fee0072d95737e1c19ad673a3c881a89de06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/992a76cf9e97e65fcbf62e72cbd57c7915f04cd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/992a76cf9e97e65fcbf62e72cbd57c7915f04cd9"}]},"branch":"refs/heads/master"},"8d09c67459694a81f78a0a64382d9bd87012ca44":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-03-30 16:08:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/914788/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/88/914788/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/3"}}},"commit":{"parents":[{"commit":"a1fe1b9cfb841d632aa582a52c106b4ebab1c159","subject":"Fix eventlet detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a1fe1b9cfb841d632aa582a52c106b4ebab1c159"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 15:54:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 16:08:04.000000000","tz":540},"subject":"Deprecate watch_log_file","message":"Deprecate watch_log_file\n\nThis feature has been broken for some time without any actual update\nto fix it. Also the pyinotify package used hasn\u0027t been updated for 9\nyears and looks badly unmaintained.\n\nRelated-Bug: #1740111\nChange-Id: If776fee0072d95737e1c19ad673a3c881a89de06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/8d09c67459694a81f78a0a64382d9bd87012ca44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/8d09c67459694a81f78a0a64382d9bd87012ca44"}]},"branch":"refs/heads/master"},"31603ac6dacd0b47ec9f02d99bfdd0a7b683119c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-03-31 05:10:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/914788/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/88/914788/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/88/914788/4"}}},"commit":{"parents":[{"commit":"a1fe1b9cfb841d632aa582a52c106b4ebab1c159","subject":"Fix eventlet detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a1fe1b9cfb841d632aa582a52c106b4ebab1c159"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 15:54:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-31 05:10:58.000000000","tz":0},"subject":"Deprecate watch_log_file","message":"Deprecate watch_log_file\n\nThis feature has been broken for some time without any actual update\nto fix it. Also the pyinotify package used hasn\u0027t been updated for 9\nyears and looks badly unmaintained.\n\nRelated-Bug: #1740111\nRelated-Bug: #2059855\nChange-Id: If776fee0072d95737e1c19ad673a3c881a89de06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/31603ac6dacd0b47ec9f02d99bfdd0a7b683119c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/31603ac6dacd0b47ec9f02d99bfdd0a7b683119c"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
