)]}'
{"id":"openstack%2Foslo.log~915158","triplet_id":"openstack%2Foslo.log~stable%2F2024.1~If7439d111c66d00a2e4cb7425f48d3d501e5a124","project":"openstack/oslo.log","branch":"stable/2024.1","topic":"bug/1946496","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-10 10:50:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7439d111c66d00a2e4cb7425f48d3d501e5a124","subject":"Fix broken reference to rate_limit_except_level","status":"MERGED","created":"2024-04-09 12:10:06.000000000","updated":"2024-04-10 10:51:56.000000000","submitted":"2024-04-10 10:50:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915158-bug/1946496","cherry_pick_of_change":914787,"cherry_pick_of_patch_set":1,"meta_rev_id":"77aebdbcb45a4466ed9e4c84c81e53ac40ba6e03","_number":915158,"virtual_id_number":915158,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-10 10:50:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-10 09:18:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-10 09:18:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-09 14:08:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-10 09:18:14.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"93c0f381dbf55ee0db1a0c958a8e7198db242ac6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-09 12:10:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d12f4cbf265288e1db2b46407ca91a398cadba48","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-09 12:10:26.000000000","message":"Topic bug/1946496-stable/2024.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"63b7f778f7d8c6a39cafbcc9a1ed51094c029290","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-09 12:10:30.000000000","message":"Topic set to bug/1946496","accounts_in_message":[],"_revision_number":1},{"id":"7b6390cfe207b33c670a106f20bc47db213a8dfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 14:08:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bde69925799743c1982e8f5353fe95b8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d2402b473ed4507aacf0fdfe51da467 : SUCCESS in 1h 38m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8f4406215da47738e44f171e7cc7df5 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a90ce3823c1f4edf94af488e022bb9c5 : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9c875d2af5d440f97bb548e7cd053c9 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a0dcc528f74b45a6f8540eab5f4994 : SUCCESS in 4m 04s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/72e84d5e7ee34f819c82cf45f6ef0a5b : SUCCESS in 1h 13m 19s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/b6ad55dce3804e708afc6d5e403bb262 : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2b57a3d9df26a76d94fc1ac7a954cd81da96447a","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-04-10 09:18:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"601bd46a7762ca867efdff832b6bcf5b0509328a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 09:18:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f860998d92b66af55f8030806c9cf28059ef01d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 10:50:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d643b0ba9ee84f80a1258fdf239d8da7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40a5eea1b4e544ba92273d8d21d03cda : SUCCESS in 1h 27m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2823cab3953a407aa9878da1a0d51509 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d92ef834ad64b858f6f22f63b6922a3 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/efd4654f17aa412c8dd15c1420142f9e : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e5f468ea344c84a3369c8146e0a01f : SUCCESS in 3m 27s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/3f24fe1070ae464f8d5fa7b519afaa98 : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":1},{"id":"236a5c4461b6abf3e5e838894f1fcb2a3e0982b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 10:50:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"77aebdbcb45a4466ed9e4c84c81e53ac40ba6e03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 10:51:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d4d04f19a0e43e98d1d788c126b5472\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e876210ae484a30ab8755ad76ec40a6 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1717f3970ecef6f33f1a83941e3b6e5f50ef4446","revisions":{"1717f3970ecef6f33f1a83941e3b6e5f50ef4446":{"kind":"REWORK","_number":1,"created":"2024-04-09 12:10:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/915158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/58/915158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/58/915158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/58/915158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/58/915158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/58/915158/1"}}},"commit":{"parents":[{"commit":"a6c4f8a9c62ae264f819716177f3da899eaf1a54","subject":"Fix eventlet detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a6c4f8a9c62ae264f819716177f3da899eaf1a54"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 15:44:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-09 12:10:06.000000000","tz":0},"subject":"Fix broken reference to rate_limit_except_level","message":"Fix broken reference to rate_limit_except_level\n\nCloses-Bug: #1946496\nCo-Authored-By: Jonathan Mills \u003cjonmills@gmail.com\u003e\nChange-Id: If7439d111c66d00a2e4cb7425f48d3d501e5a124\n(cherry picked from commit 4bd77b692370dcaebab18110cd4bf50e15346584)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1717f3970ecef6f33f1a83941e3b6e5f50ef4446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1717f3970ecef6f33f1a83941e3b6e5f50ef4446"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
