)]}'
{"id":"openstack%2Foslo.log~915400","triplet_id":"openstack%2Foslo.log~stable%2F2023.2~If7439d111c66d00a2e4cb7425f48d3d501e5a124","project":"openstack/oslo.log","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-04-19 13:26:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7439d111c66d00a2e4cb7425f48d3d501e5a124","subject":"Fix broken reference to rate_limit_except_level","status":"MERGED","created":"2024-04-10 09:25:53.000000000","updated":"2024-04-19 13:27:21.000000000","submitted":"2024-04-19 13:26:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915400","meta_rev_id":"6676122f547262d4a09227fa602ee5dba30700d2","_number":915400,"virtual_id_number":915400,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-19 13:26:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-04-19 12:28:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-04-19 12:28:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-10 09:25:53.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-04-10 10:56:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-19 12:28:27.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"7bab3f4caa2373fec8df8cf3895524013ad46ffc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-04-10 09:25:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"259a2dc84fb458a9f6b877d6dca122d4c536bca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 10:56:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df740e9352204882bf9ca22b02e00426\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5d18ba6ea5a46b9b06af874e6b46ba5 : SUCCESS in 1h 25m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ff62bcf86f4a5da3502be2f44133cd : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24047d45dbac4ba4ac750fd19f4c75a2 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d7088c6ff504f47b476bfe9d452cc9e : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2fa4476c4d1d48c4b6af23f71bb6e522 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/112ca7c8739c4643925d0044734a7e84 : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a980956c721848e1a5d8e1244b90ba19 : SUCCESS in 2m 54s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/16e0cc3df4214338b713a8317f22e8ba : SUCCESS in 58m 25s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/d7e8243e486c4926ae7ec48b835720d0 : SUCCESS in 37m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b7825a340c55a050a465dc6247d5d1410ce44fe4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 12:28:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"06c16b68547f3b28843fe400602ec059df82bd41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 12:28:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"193a9e2ee97ed242e9bfcf5ab7e73c2a0a5a6448","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 13:26:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f5acb9233604e4c8ada971fc79f9327\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5397a2f5cb934631a9063916d06843a6 : SUCCESS in 55m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1203887b06fd4d27842cce418a85999a : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/032fa7ac7c5342c682b0918035b6b0f5 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b49331be6b3d42debebdf99ddd008939 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/daa99de7dd994642aa257552d6a75179 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acdb68ef8ef847788eb0c5ba8185374a : SUCCESS in 2m 43s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/4db42516bb22421b81178fb22c411a7e : SUCCESS in 31m 58s","accounts_in_message":[],"_revision_number":1},{"id":"6d50f7261092ec273218e91392052cd30dbfc4fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 13:26:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6676122f547262d4a09227fa602ee5dba30700d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 13:27:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb5c05c560b946f7a3274939786c5e54\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd5e189a774d4804a1c4d4ed0c947e76 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"573a434ace71ea97a8f76791690d014e70db86d5","revisions":{"573a434ace71ea97a8f76791690d014e70db86d5":{"kind":"REWORK","_number":1,"created":"2024-04-10 09:25:53.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/00/915400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/00/915400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/915400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/915400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/915400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/00/915400/1"}}},"commit":{"parents":[{"commit":"287e138adf66c28b7c127dce99a8af05df00225e","subject":"Fix eventlet detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/287e138adf66c28b7c127dce99a8af05df00225e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 15:44:41.000000000","tz":540},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-04-10 09:25:41.000000000","tz":120},"subject":"Fix broken reference to rate_limit_except_level","message":"Fix broken reference to rate_limit_except_level\n\nCloses-Bug: #1946496\nCo-Authored-By: Jonathan Mills \u003cjonmills@gmail.com\u003e\nChange-Id: If7439d111c66d00a2e4cb7425f48d3d501e5a124\n(cherry picked from commit 4bd77b692370dcaebab18110cd4bf50e15346584)\n(cherry picked from commit 1717f3970ecef6f33f1a83941e3b6e5f50ef4446)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/573a434ace71ea97a8f76791690d014e70db86d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/573a434ace71ea97a8f76791690d014e70db86d5"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
