)]}'
{"id":"openstack%2Foslo.log~915802","triplet_id":"openstack%2Foslo.log~stable%2F2024.1~I125f7a8fa12f4d17af70987c20e5dedb14232d7b","project":"openstack/oslo.log","branch":"stable/2024.1","topic":"bug/1995514","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-28 13:01:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I125f7a8fa12f4d17af70987c20e5dedb14232d7b","subject":"Apply eventlet workaround only once","status":"MERGED","created":"2024-04-15 12:06:50.000000000","updated":"2024-05-28 13:02:43.000000000","submitted":"2024-05-28 13:01:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915802-bug/1995514","cherry_pick_of_change":914790,"cherry_pick_of_patch_set":2,"meta_rev_id":"a0131e95fae923d286cd35cc615dc39400dfa180","_number":915802,"virtual_id_number":915802,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-28 13:01:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-28 11:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-28 11:46:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-15 13:42:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-28 11:46:17.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"3035f317212552f330c6bcb58d3f30cc8416b432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 12:06:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"20a14a61e3aef51af575c8eca72eb3ed760887c4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 12:07:12.000000000","message":"Topic bug/1995514-stable/2024.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"b8a90a574cf40ed8051e66d223f9c53e18859199","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 12:07:18.000000000","message":"Topic set to bug/1995514","accounts_in_message":[],"_revision_number":1},{"id":"78d13371811112ca5bab528990a6182b14065412","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 13:42:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00172504069f4b56a26eb8adc4d4e7a7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3895b02fd314169a4933b413dbe045c : SUCCESS in 1h 29m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f472377049734845b559d6ad07ab0c4e : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aeed2419c7644c61addda3debd733a4a : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a5ffaabca22401fb1f6205a52062143 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cbe713937024bbd81628e909c1e51cf : SUCCESS in 4m 07s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/d66d59d762754f94822e9bb3ffb50973 : SUCCESS in 1h 05m 33s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/426ef40681384766aadf8b60e88831ca : SUCCESS in 31m 10s","accounts_in_message":[],"_revision_number":1},{"id":"62ebb30ea0daf76c2d6c1ab5e419a19e4c690e06","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-05-28 11:46:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e354f9bc16d44bfa7d534b98138861ac9cb3d04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 11:46:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec63ddf35f155685c77737859a6f01a3fdde42be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 13:01:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9d674becd9842939b3f764be89c3a0b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6077f2bf99974995bd6be284fdaebac7 : SUCCESS in 1h 14m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16d58b6c4fb6488eb4fcb0cef5fff85e : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac788b6403504c8f81e0092ba814d3d2 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/639cad99b5fe4911a3ff262e1953cd02 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53a28b924fc4dd7949f76bae4f5853e : SUCCESS in 3m 21s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/06a81031adf543629a7a151a587d5910 : SUCCESS in 23m 41s","accounts_in_message":[],"_revision_number":1},{"id":"06aa2c6929729da30a60686da47f1048df9b55c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 13:01:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a0131e95fae923d286cd35cc615dc39400dfa180","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 13:02:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/463ae61883b1496bbe88577c0c8e496e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd20b19eeb224af58c96f84b72d107f9 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11a9976e52faa97645d77ff56030798d16807998","revisions":{"11a9976e52faa97645d77ff56030798d16807998":{"kind":"REWORK","_number":1,"created":"2024-04-15 12:06:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/915802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/02/915802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/02/915802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/02/915802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/02/915802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/02/915802/1"}}},"commit":{"parents":[{"commit":"1717f3970ecef6f33f1a83941e3b6e5f50ef4446","subject":"Fix broken reference to rate_limit_except_level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1717f3970ecef6f33f1a83941e3b6e5f50ef4446"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 16:23:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:06:50.000000000","tz":0},"subject":"Apply eventlet workaround only once","message":"Apply eventlet workaround only once\n\n... to avoid issues caused by setup method called multiple times.\n\nCloses-Bug: #1995514\nChange-Id: I125f7a8fa12f4d17af70987c20e5dedb14232d7b\n(cherry picked from commit 142fde8224c8994b94518803fe5005cdeec6a37e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/11a9976e52faa97645d77ff56030798d16807998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/11a9976e52faa97645d77ff56030798d16807998"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
